)]}'
{"id":"openstack%2Fnetworking-sfc~799838","triplet_id":"openstack%2Fnetworking-sfc~master~I2e31be14f7b768b563d1c6f4babb08ec6c7ca9a2","project":"openstack/networking-sfc","branch":"master","topic":"bug/1934900","hashtags":[],"change_id":"I2e31be14f7b768b563d1c6f4babb08ec6c7ca9a2","subject":"[functional] Remove duplicate code from parent test class","status":"MERGED","created":"2021-07-07 14:01:43.000000000","updated":"2021-08-31 16:24:08.000000000","submitted":"2021-08-31 16:22:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799838-bug/1934900","meta_rev_id":"078ab1e715ab7016e2197b4620aa5dde382e0bb5","_number":799838,"virtual_id_number":799838,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-31 16:22:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2021-07-09 04:01:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2021-08-31 15:05:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2021-08-25 12:36:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":1,"date":"2021-08-31 15:05:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-07 15:05:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-09 04:01:46.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-08-24 16:02:08.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-08-24 16:02:08.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-08-26 13:11:06.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b600e49a0acc571e100c57c1c43dad117c7b0228","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-07-07 14:01:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ea6b6dd79410b7b66671c4367b2c1c190764e2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 15:05:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3c5076deb9f494da6330a4dfd7eed8a : SUCCESS in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c58c4c7266a434ba12022bac583e873 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eeed85c2d34d4369a71c8d98d642dcdc : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2cd60fd342f4b8b94fd5478bc446155 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88d59f6c99e5491099ae742cde5e324a : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8468eb2dc2046f591c4b9b829c6280e : SUCCESS in 6m 18s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/ea6263d940974087976d39aa452c74d5 : SUCCESS in 20m 38s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/c76b6cba6bc84bd2bc970bccb4e1253d : SUCCESS in 1h 01m 42s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/ac1e4e9958584e438566eb971464b7b0 : FAILURE in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9880709a263b95808e35c4082979d831afb1192","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-07-09 04:01:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b91a8d2db062754636c3924a6cfc94cea647370c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-08-25 12:36:29.000000000","message":"Patch Set 1: Code-Review+2\n\nGood catch","accounts_in_message":[],"_revision_number":1},{"id":"07fd9ec90d25b86d780706b7d12141202be93bd8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-08-31 15:05:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05eea5e0371591d7e6ca876a27a1e95b9e180364","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 15:06:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0c7863204e4c3fecceddacc9ef8384ab0aacb665","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 16:22:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a6fecd1619145fea13e163fc7d2629d : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ceac09a852c4c75a9dce42b804b3376 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2b002e3c44d43c294bc084623058c35 : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4e14a063c9047a892eda7164c13105b : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2aa5ec5cdb4e34ab4b0a7442f90d42 : SUCCESS in 9m 09s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/5836e386c65a412b8db8aa7e64aba90f : SUCCESS in 25m 15s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/68d3b2a827d7430e85d787986b629053 : SUCCESS in 1h 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4dd975af7cde8c2c2e409a4c560c229b4c9f791b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 16:22:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"078ab1e715ab7016e2197b4620aa5dde382e0bb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 16:24:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22541351302947f0acdc957a2e48b1f5 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f211d5d2ba2bfb482de84b1884df6e846277b952","revisions":{"f211d5d2ba2bfb482de84b1884df6e846277b952":{"kind":"REWORK","_number":1,"created":"2021-07-07 14:01:43.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/38/799838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/38/799838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/38/799838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/38/799838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/38/799838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/38/799838/1"}}},"commit":{"parents":[{"commit":"e01048cc521c2c16f36cea23eef8331c6bae0904","subject":"Use assertCountEqual instead of assertItemsEqual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e01048cc521c2c16f36cea23eef8331c6bae0904"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-07-07 13:52:05.000000000","tz":120},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2021-07-07 13:59:32.000000000","tz":120},"subject":"[functional] Remove duplicate code from parent test class","message":"[functional] Remove duplicate code from parent test class\n\nThe server start functional test had code identical to parent class\nTestPluginWorker that can be dropped. It also broke in networking-sfc\nafter neutron https://review.opendev.org/c/openstack/neutron/+/793899\nwas merged\n\nChange-Id: I2e31be14f7b768b563d1c6f4babb08ec6c7ca9a2\nCloses-Bug: #1934900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f211d5d2ba2bfb482de84b1884df6e846277b952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f211d5d2ba2bfb482de84b1884df6e846277b952"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
