)]}'
{"id":"openstack%2Fnetworking-sfc~844290","triplet_id":"openstack%2Fnetworking-sfc~master~If769e89896d85cc1626fa111f546d935da21a590","project":"openstack/networking-sfc","branch":"master","topic":"fix-multinode-job","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-09 07:57:32.000000000","reason":"Rodolfo Alonso replied on the change"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-06-09 10:56:07.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-06-02 13:00:30.000000000","reason":"Bernard Cafarelli replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-06-09 09:27:18.000000000","reason":"Slawek Kaplonski replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-06-09 10:56:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If769e89896d85cc1626fa111f546d935da21a590","subject":"Fix networking-sfc-tempest-multinode job","status":"MERGED","created":"2022-06-01 13:07:19.000000000","updated":"2022-06-09 10:58:21.000000000","submitted":"2022-06-09 10:56:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"844290-fix-multinode-job","meta_rev_id":"ec0f4666757aa0a1b20ea1d416dda5dfc5f2092b","_number":844290,"virtual_id_number":844290,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-09 10:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-06-09 07:57:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-06-02 13:00:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-06-09 09:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-06-09 09:27:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 14:06:07.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-06-01 14:34:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 10:45:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-06-02 10:45:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-06-02 10:45:35.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"0431e9185d16a2b2ec6752653570e2daadc60fab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-01 13:07:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a094f4ff9cc286c72d56431c753515f212eba0b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-01 14:09:11.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7198c88b8629432c0958f007d70c1c9d015d90c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 14:34:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2e14010677a4d8f97d52247d87f911b : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc7182dacd0b45b1a9173cc2bffe3fd7 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/506c09fe278043f99e3a134d3ed979dd : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/736aba3f33b948f9955d9800df7c2c88 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e27b9816888248318b5150e6dc439bb2 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6210b00c98214dd3b426e6fa466d7cec : SUCCESS in 7m 21s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/616f09ad44454fb59f4b609d1f1daa65 : SUCCESS in 1h 23m 58s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/78167fe14af74b008dade9dff0c75e56 : FAILURE in 53m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f56256d24a1e7548b4d986440056d4807639e6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-01 15:11:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f10e047e5d6080f5f74c9f98908e842a728eb04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 16:23:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/89d2c53be6644721aff535f0c990ee1d : FAILURE in 1h 05m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa42cd23474962b3083654c9d0f2edc90b2df497","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-02 04:36:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"11dc6a9617a4e19b5d9bb12e9b373114fca1d2e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 06:01:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/579478a75b8947de94e6e116cbf28b25 : FAILURE in 1h 23m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ddb527c72be6d9a2cb004b9bff63a8fbd063c6c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-02 06:49:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c0a44608e29e1997b0c886ff5e1da35eabec0dde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 08:13:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cba87a46ba8341f3ae2f7333c8fb2521 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64181e36452849709a3cad527cab3edc : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba4a1757fb284c44989b91337b615e42 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d848b6e28af466094cc8df2fe4c90e7 : SUCCESS in 6m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfbf2e6c1bbd485b8d814f380be0c02b : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/868197c97e1141e2b9ca23fc12bc8116 : SUCCESS in 7m 18s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/0561b5dcb60542d59fdd92d89b80001d : SUCCESS in 14m 21s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/68b80be60a9e438484dc01bd3fd80075 : SUCCESS in 1h 09m 08s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/dc3e535df1c94791b41ad8af0c94426c : FAILURE in 1h 21m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3225ac7fd3e3f80c8f73e0a9a51af17fa9c08e18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-02 08:30:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0dcfaba57089c27dffb88a1ae494d02c3b0134de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 10:16:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc3f3f0dbe4844e08cb975ca2ca33e73 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0825ef8116c340bf9be3ad9864d9f34b : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cb18b3d41b340adacf206be0be60fbf : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ba28cc1c8c1432aa6224f1a8a3bd3ed : SUCCESS in 7m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/365e253c15374a7b848cddb68ff5c00b : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1035209a9ca94ad38f604aa7028e4d1c : SUCCESS in 7m 47s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/485b6b9b474b492ca4081b0960dae953 : SUCCESS in 22m 23s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/ca5a3b17e9574276a84c259ad49bf32a : SUCCESS in 1h 44m 50s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/a3b1464f433b4a91a6825f7690f595aa : SUCCESS in 1h 37m 24s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4bd2ce778467074da5077f9b67ca4335f0bb9665","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-02 13:00:30.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"73422c42a3da8586dfe1ecc86d213318257a387d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-09 07:57:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"497b49fb470c88089ae5565ccb36f15179f61c0f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-09 09:27:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"aa23b9774a32d299740a4986f303203229e64dff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 09:28:02.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f41b210c1ff4f548f47edb897439c29269dd5d60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 10:56:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6709e81eaefa41db85dd79121d1966d2 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562b50d8cd4b434baa6f99f6a358af29 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c3cee7181f84473a8ffe65fbde5e904 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adc72b31f09647edb9809f71e02deb70 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd68c9bd50a493bb1b54f0f4e87ed7f : SUCCESS in 8m 34s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/acb4d5f3c77441d2840ed8b6d6d58c82 : SUCCESS in 18m 10s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/7dac062e272249cfa2a5a24d9c43cce0 : SUCCESS in 1h 25m 32s","accounts_in_message":[],"_revision_number":5},{"id":"727ec57adc7da94160c00b7a9c7cdbc7086d259b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 10:56:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"ec0f4666757aa0a1b20ea1d416dda5dfc5f2092b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 10:58:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4740f86d9241a9a9ce900351d30895 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"161c6ee578e2397abf520e227d74dfd61b99fdfa","revisions":{"0270b5a29097cf2b757a86a616c086c8872a5bc7":{"kind":"REWORK","_number":1,"created":"2022-06-01 13:07:19.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/90/844290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/90/844290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/1"}}},"commit":{"parents":[{"commit":"9202e42a0aca15e6f63e6687944e40463db863a3","subject":"[sqlalchemy-20] Add Context for all SQL transactions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/9202e42a0aca15e6f63e6687944e40463db863a3"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-01 12:41:12.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-01 12:41:14.000000000","tz":330},"subject":"Fix networking-sfc-tempest-multinode job","message":"Fix networking-sfc-tempest-multinode job\n\ncinder was enabled while c-api not leading to\ndeployment failure, removing cinder from enabled\nservices list to fix this as the service is not\nneeded in this job.\n\nChange-Id: If769e89896d85cc1626fa111f546d935da21a590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0270b5a29097cf2b757a86a616c086c8872a5bc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0270b5a29097cf2b757a86a616c086c8872a5bc7"}]},"branch":"refs/heads/master"},"f07d5df725cfa09e9ec828eb4fd0717dc1069a8d":{"kind":"REWORK","_number":2,"created":"2022-06-01 15:11:06.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/90/844290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/90/844290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/2"}}},"commit":{"parents":[{"commit":"ddb298c77296a7e780e33a75c90e4a5252169617","subject":"[sqlalchemy-20] Add Context for all SQL transactions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ddb298c77296a7e780e33a75c90e4a5252169617"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-01 12:41:12.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-01 15:07:48.000000000","tz":330},"subject":"[WIP] Fix networking-sfc-tempest-multinode job","message":"[WIP] Fix networking-sfc-tempest-multinode job\n\ncinder was enabled while c-api not leading to\ndeployment failure, removing cinder from enabled\nservices list to fix this as the service is not\nneeded in this job.\nAlso disable ovn services and enable required agents,\nplugins.\n\nChange-Id: If769e89896d85cc1626fa111f546d935da21a590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f07d5df725cfa09e9ec828eb4fd0717dc1069a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f07d5df725cfa09e9ec828eb4fd0717dc1069a8d"}]},"branch":"refs/heads/master"},"a898847d38a3ecb93cd5426ce62b94aca0f3ac9a":{"kind":"REWORK","_number":3,"created":"2022-06-02 04:36:27.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/90/844290/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/90/844290/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/3"}}},"commit":{"parents":[{"commit":"ddb298c77296a7e780e33a75c90e4a5252169617","subject":"[sqlalchemy-20] Add Context for all SQL transactions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ddb298c77296a7e780e33a75c90e4a5252169617"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-01 12:41:12.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-02 04:36:25.000000000","tz":330},"subject":"[WIP] Fix networking-sfc-tempest-multinode job","message":"[WIP] Fix networking-sfc-tempest-multinode job\n\ncinder was enabled while c-api not leading to\ndeployment failure, removing cinder from enabled\nservices list to fix this as the service is not\nneeded in this job.\nAlso disable ovn services and enable required agents,\nplugins.\n\nChange-Id: If769e89896d85cc1626fa111f546d935da21a590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/a898847d38a3ecb93cd5426ce62b94aca0f3ac9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/a898847d38a3ecb93cd5426ce62b94aca0f3ac9a"}]},"branch":"refs/heads/master"},"88e95dc49fb106fde4cf27301736f4b5ca47a792":{"kind":"REWORK","_number":4,"created":"2022-06-02 06:49:00.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/90/844290/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/90/844290/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/4"}}},"commit":{"parents":[{"commit":"ddb298c77296a7e780e33a75c90e4a5252169617","subject":"[sqlalchemy-20] Add Context for all SQL transactions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ddb298c77296a7e780e33a75c90e4a5252169617"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-01 12:41:12.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-02 06:48:13.000000000","tz":330},"subject":"Fix networking-sfc-tempest-multinode job","message":"Fix networking-sfc-tempest-multinode job\n\ncinder was enabled while c-api not leading to\ndeployment failure, removing cinder from enabled\nservices list to fix this as the service is not\nneeded in this job and fix job config for subnode\nas per openvswitch agent.\n\nAlso pass plugin context to l2pop.db.get_agent_by_host\nas without it it fails with:-\nAttributeError: \u0027PortChainContext\u0027 object has no attribute \u0027session\u0027\n\nChange-Id: If769e89896d85cc1626fa111f546d935da21a590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/88e95dc49fb106fde4cf27301736f4b5ca47a792"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/88e95dc49fb106fde4cf27301736f4b5ca47a792"}]},"branch":"refs/heads/master"},"161c6ee578e2397abf520e227d74dfd61b99fdfa":{"kind":"REWORK","_number":5,"created":"2022-06-02 08:30:29.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/90/844290/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/90/844290/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/90/844290/5"}}},"commit":{"parents":[{"commit":"ddb298c77296a7e780e33a75c90e4a5252169617","subject":"[sqlalchemy-20] Add Context for all SQL transactions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ddb298c77296a7e780e33a75c90e4a5252169617"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-01 12:41:12.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-02 08:30:19.000000000","tz":330},"subject":"Fix networking-sfc-tempest-multinode job","message":"Fix networking-sfc-tempest-multinode job\n\ncinder was enabled while c-api not leading to\ndeployment failure, removing cinder from enabled\nservices list to fix this as the service is not\nneeded in this job and fix job config for subnode\nas per openvswitch agent.\n\nAlso pass plugin context to l2pop.db.get_agent_by_host\nas without it it fails with:-\nAttributeError: \u0027PortChainContext\u0027 object has no attribute \u0027session\u0027\n\nChange-Id: If769e89896d85cc1626fa111f546d935da21a590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/161c6ee578e2397abf520e227d74dfd61b99fdfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/161c6ee578e2397abf520e227d74dfd61b99fdfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
