)]}'
{"id":"openstack%2Fnetworking-sfc~848223","triplet_id":"openstack%2Fnetworking-sfc~master~I650684473a9f6307c7e4d11f4764e3afeb99b05f","project":"openstack/networking-sfc","branch":"master","topic":"drop-l-c","attention_set":{},"removed_from_attention_set":{"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-07-04 01:46:00.000000000","reason":"Change was abandoned"},"33436":{"account":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"last_update":"2022-07-04 01:45:55.000000000","reason":"\u003cGERRIT_ACCOUNT_33436\u003e replied on the change","reason_account":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"}}},"hashtags":[],"change_id":"I650684473a9f6307c7e4d11f4764e3afeb99b05f","subject":"Drop lower-constraints.txt and its testing","status":"ABANDONED","created":"2022-06-30 07:43:05.000000000","updated":"2022-07-04 01:46:00.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6228fce530ea08ff3fdcedc37cf5e657e2697fbf","_number":848223,"virtual_id_number":848223,"owner":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-30 10:27:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":-1,"date":"2022-07-01 10:13:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-30 10:27:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-01 10:13:42.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"889ed78bbbde970a86a147a4fcffffa489a8a340","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2022-06-30 07:43:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39bff556bf9ca7ac4e0e995c48bb540557fd9aa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2022-06-30 07:44:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3332139cbcfcb6aea2b864d32e4e72b23367b292","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2022-06-30 07:47:08.000000000","message":"Topic set to drop-l-c","accounts_in_message":[],"_revision_number":2},{"id":"c3b8392f9f82067c1e6d8fbe9ed993e2579a973e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2022-06-30 08:29:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8abc4e68c610844b8adfc50affaaec26d10c518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 10:27:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b389908d604404b7be702748ab7867 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e0fcf4c05c9464fab20a127ab19131d : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28471518b2844fc2939aceea83d7a998 : SUCCESS in 8m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d6e027d4f3504b8caad57c75a452847e : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd7eabe47d8749f0a71698a77f270207 : SUCCESS in 8m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da1e7cb5abcf4da3bff841dd0a7f935a : SUCCESS in 2m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/089c3c8e9a96472a9ad7d57b1caf15a5 : SUCCESS in 3m 03s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/f0a447f3b4474661bcf9ecedb233106c : SUCCESS in 25m 51s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/e4e57264dbd34d9c9064707b1f78b5af : SUCCESS in 1h 32m 51s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/205ce3b196da4b379edf880c24e178ba : SUCCESS in 1h 50m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f8eb4bdf14054f0384b5e7a4122667decfaacfb","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-07-01 10:13:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"40a893abad9f9ba915846e61b06ab3025db59e15","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2022-07-04 01:45:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6228fce530ea08ff3fdcedc37cf5e657e2697fbf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"date":"2022-07-04 01:46:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"45b3e20476bc690bdc1f1d3b385880401fd7bbff","revisions":{"1ad4b4b1b3b9c34d1a13c0a190c7fc28c1067dc5":{"kind":"REWORK","_number":1,"created":"2022-06-30 07:43:05.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/23/848223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/23/848223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/1"}}},"commit":{"parents":[{"commit":"ddb298c77296a7e780e33a75c90e4a5252169617","subject":"[sqlalchemy-20] Add Context for all SQL transactions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ddb298c77296a7e780e33a75c90e4a5252169617"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-30 07:42:24.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-30 07:42:38.000000000","tz":480},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I650684473a9f6307c7e4d11f4764e3afeb99b05f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/1ad4b4b1b3b9c34d1a13c0a190c7fc28c1067dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/1ad4b4b1b3b9c34d1a13c0a190c7fc28c1067dc5"}]},"branch":"refs/heads/master"},"32824bdbbf0f095d8415dce70b6d60a48e430752":{"kind":"REWORK","_number":2,"created":"2022-06-30 07:44:41.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/23/848223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/23/848223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/2"}}},"commit":{"parents":[{"commit":"ddb298c77296a7e780e33a75c90e4a5252169617","subject":"[sqlalchemy-20] Add Context for all SQL transactions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ddb298c77296a7e780e33a75c90e4a5252169617"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-30 07:42:24.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-30 07:44:29.000000000","tz":480},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I650684473a9f6307c7e4d11f4764e3afeb99b05f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/32824bdbbf0f095d8415dce70b6d60a48e430752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/32824bdbbf0f095d8415dce70b6d60a48e430752"}]},"branch":"refs/heads/master"},"45b3e20476bc690bdc1f1d3b385880401fd7bbff":{"kind":"REWORK","_number":3,"created":"2022-06-30 08:29:20.000000000","uploader":{"_account_id":33436,"name":"YuehuiLei","display_name":"leiyuehui","email":"leiyuehui-s@inspur.com","username":"leiyuehui"},"ref":"refs/changes/23/848223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/23/848223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/23/848223/3"}}},"commit":{"parents":[{"commit":"ddb298c77296a7e780e33a75c90e4a5252169617","subject":"[sqlalchemy-20] Add Context for all SQL transactions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ddb298c77296a7e780e33a75c90e4a5252169617"}]}],"author":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-30 07:42:24.000000000","tz":480},"committer":{"name":"leiyuehui","email":"leiyuehui-s@inspur.com","date":"2022-06-30 08:29:07.000000000","tz":480},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I650684473a9f6307c7e4d11f4764e3afeb99b05f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/45b3e20476bc690bdc1f1d3b385880401fd7bbff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/45b3e20476bc690bdc1f1d3b385880401fd7bbff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
