)]}'
{"id":"openstack%2Fnetworking-sfc~868168","triplet_id":"openstack%2Fnetworking-sfc~master~I54ff765d1f839232c85ab01556be03d95f135bd5","project":"openstack/networking-sfc","branch":"master","topic":"drop-lib-neutron","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-12-20 15:37:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I54ff765d1f839232c85ab01556be03d95f135bd5","subject":"Remove note about migration from lib/neutron-legacy to lib/neutron","status":"MERGED","created":"2022-12-20 10:12:41.000000000","updated":"2022-12-20 15:39:47.000000000","submitted":"2022-12-20 15:37:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868168-drop-lib-neutron","meta_rev_id":"076a504ec0fcc3c48cb8c111c8147a7475cc9644","_number":868168,"virtual_id_number":868168,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-20 15:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-12-20 14:19:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-12-20 14:54:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-12-20 12:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-12-20 14:19:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-20 12:06:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-20 12:11:22.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-12-20 14:19:24.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-12-20 14:54:24.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"e8d798d82d4e7124adc9f9d8e22652a358b6bc0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-12-20 10:12:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35a261e479364bd69b087f47b57a42193ae10251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 12:06:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dbe8dab5b094920aae21e0ef2355be2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2fb65855dce481490b5a36af7f5141a : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc66ae12903140c798593b8b4886495d : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df12961e23264994b376af4fad55fedf : SUCCESS in 7m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed1ed93a988642389dd5160a35d02019 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a7d4e05930b434888f0b1d6f35ad232 : SUCCESS in 7m 09s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/b5814cbdf03844689e86e1e7644c4751 : SUCCESS in 1h 20m 42s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/9a099ddae3024b1e8b0cf79b32e5a0d1 : SUCCESS in 1h 46m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ba580dd60d031a896cda403350d67587d5acf79","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-12-20 12:11:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1011a70e32486c72353970d8e790599f84b53b77","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-12-20 14:19:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbf0c3c31dcccc22a566951ca1f7671fa64a7a53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 14:20:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f885e9ef7e6937949b1c348372d90c9a685dd7ca","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-12-20 14:54:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cd72d7da46a6425bc62d5bafdea65734561c4c13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 15:37:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eca63105be4b487fa79e895a76a9a947\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0548a061de94531bcaa7f3b40a46c56 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c8d21f3fd6a48e59349470fcee0c3ba : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/326a82196dd54f87b242995b2fd1e24e : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d54036446f439baeb19635612e512e : SUCCESS in 6m 47s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/aeaa9cb49e9d42aa82b2d9f407472dab : SUCCESS in 1h 16m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e05e9a069d01727f3d8c24672571cf86b5d8e8bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 15:37:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"076a504ec0fcc3c48cb8c111c8147a7475cc9644","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 15:39:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d855ecbb4a0d4244b71df82f6bbb1113\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d167715bdd417997d79e62ca72bc9b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa1dbcd1b379ab755049173e02f5df5850c42b09","revisions":{"aa1dbcd1b379ab755049173e02f5df5850c42b09":{"kind":"REWORK","_number":1,"created":"2022-12-20 10:12:41.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/68/868168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/68/868168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/868168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/868168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/868168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/68/868168/1"}}},"commit":{"parents":[{"commit":"f3de0806c8bac71605edab30fffefee0380edf1f","subject":"Cleanup tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f3de0806c8bac71605edab30fffefee0380edf1f"}]}],"author":{"name":"Sławek Kapłoński","email":"skaplons@redhat.com","date":"2022-12-20 10:12:22.000000000","tz":60},"committer":{"name":"Sławek Kapłoński","email":"skaplons@redhat.com","date":"2022-12-20 10:12:40.000000000","tz":60},"subject":"Remove note about migration from lib/neutron-legacy to lib/neutron","message":"Remove note about migration from lib/neutron-legacy to lib/neutron\n\nIn [1] we finally got rid of the unfinished lib/neutron module and kept\nonly lib/neutron-legacy. It\u0027s renamed to lib/neutron now and it\u0027s the\nonly neutron related module in Devstack.\nSo this patch removes left over todo comments about things to do when\nmigration to the new lib/neutron module will be finished.\n\n[1] https://review.opendev.org/c/openstack/devstack/+/865014\n\nChange-Id: I54ff765d1f839232c85ab01556be03d95f135bd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/aa1dbcd1b379ab755049173e02f5df5850c42b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/aa1dbcd1b379ab755049173e02f5df5850c42b09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
