)]}'
{"id":"openstack%2Fnetworking-sfc~869551","triplet_id":"openstack%2Fnetworking-sfc~master~I45b07cd9a7c29be1b579d0def64e2235336a9244","project":"openstack/networking-sfc","branch":"master","topic":"tox4","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-01-12 14:56:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I45b07cd9a7c29be1b579d0def64e2235336a9244","subject":"tox4: Add ignore_basepython_conflict \u003d True to tox.ini","status":"MERGED","created":"2023-01-09 10:41:31.000000000","updated":"2023-01-12 14:57:49.000000000","submitted":"2023-01-12 14:56:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869551-tox4","meta_rev_id":"c7f6c9e2859573c17c33aef4b5f23ef94319eff8","_number":869551,"virtual_id_number":869551,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 14:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":2,"date":"2023-01-09 12:29:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2023-01-10 04:41:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":2,"date":"2023-01-12 13:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":1,"date":"2023-01-12 13:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-09 12:22:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-09 12:29:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-01-10 04:41:43.000000000","updated_by":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"reviewer":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"state":"REVIEWER"},{"updated":"2023-01-12 13:31:59.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"59687bbcc84af2b60e84c16aa3d5eb921b9f7c31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-01-09 10:41:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d4d3c1c38fb6cc76a7acf746a36a9953baed1b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-09 12:22:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ffb4dbb654a4e8cb8e574b78959b8bd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51d3a0f0a47f4d61a5212422332bc7ed : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e236cbb669ab489283d846da4164ba30 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abab1d50498f40abbd3cd6754bcec4c4 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56efc17597ec4537ab0215bbf0fb8021 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/866cfd8a5abd46f7a457c27225275807 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e364ca7901240aa937fcba97d0b4c78 : SUCCESS in 3m 28s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/1a8ac96ed2824ceaa89c9b8949095abd : SUCCESS in 25m 35s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/bb20d6f20f7643fe973b51668f32abb1 : SUCCESS in 1h 35m 57s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/d2bfe45274404caca9f33d6c5806f805 : SUCCESS in 1h 35m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f671b96521f1c46c78a65fd45697c7498fa82da8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-01-09 12:29:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9e415f7893ad1fcbab368f62b9d170c3163f772d","author":{"_account_id":32102,"name":"Manpreet Kaur","email":"kaurmanpreet2620@gmail.com","username":"manpreet"},"date":"2023-01-10 04:41:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9c963dcd0bf08cbecb18f2241ef7e70a8dd7afce","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2023-01-12 13:31:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"825f7f39c34594a010e3ad8c1cfff6212ec91534","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 13:32:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d49c709231bb6a0e063d88736802f11bfa47067d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:56:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d55b2033bbe0402b8d40d00cfca1c7d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c9aae249e347158570c3266ac062d2 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb2fcac270b14dc7a78c81ed74e8a3b8 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0d3f29a1d9f4541974b3990dc1f1d69 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34a124cc75c74979b2ea06fe58d43f0b : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39f37bb7a64d446da91b13aa1a3c1947 : SUCCESS in 3m 44s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/d896437563564a359774ca747746dfde : SUCCESS in 17m 14s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/e67173296ed9400991a910782adf548b : SUCCESS in 1h 21m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a33fb6821072b037d3139e1d16a0656c80345660","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:56:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c7f6c9e2859573c17c33aef4b5f23ef94319eff8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 14:57:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eec82fdb2954c37847e350cfc427472\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8025e43607a44e1788af029a069f9a26 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/817963976b0740f787727fa8899fd8f8 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"541be2ece5d4f94ba8a8f2ef904556b8f1dda7d6","revisions":{"541be2ece5d4f94ba8a8f2ef904556b8f1dda7d6":{"kind":"REWORK","_number":1,"created":"2023-01-09 10:41:31.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/51/869551/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/51/869551/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/51/869551/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/51/869551/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/51/869551/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/51/869551/1"}}},"commit":{"parents":[{"commit":"16f631dc3b07e94e7bfaf9aab7632caad739db2a","subject":"Fix tox.ini for tox4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/16f631dc3b07e94e7bfaf9aab7632caad739db2a"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-01-09 10:37:37.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-01-09 10:41:06.000000000","tz":60},"subject":"tox4: Add ignore_basepython_conflict \u003d True to tox.ini","message":"tox4: Add ignore_basepython_conflict \u003d True to tox.ini\n\nWith latest tox (\u003e4.0.15) python version conflicted with base\npython. Adding ignore_basepython_conflict \u003d True solves this\nissue.\n\nChange-Id: I45b07cd9a7c29be1b579d0def64e2235336a9244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/541be2ece5d4f94ba8a8f2ef904556b8f1dda7d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/541be2ece5d4f94ba8a8f2ef904556b8f1dda7d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
