)]}'
{"id":"openstack%2Fnetworking-sfc~879336","triplet_id":"openstack%2Fnetworking-sfc~master~I16a823d9274f4ec5290c1e2d5d009d02f35623c0","project":"openstack/networking-sfc","branch":"master","topic":"bug/2004265","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-04-12 12:57:52.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-04-17 15:58:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I16a823d9274f4ec5290c1e2d5d009d02f35623c0","subject":"Add neutron and neutron-lib projects to SQLAlchemy main branch job","status":"MERGED","created":"2023-04-03 12:27:46.000000000","updated":"2023-04-17 16:01:12.000000000","submitted":"2023-04-17 15:58:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"879336-bug/2004265","meta_rev_id":"faed7a097b3f5252d5f80ecc5ee98f9545b00af8","_number":879336,"virtual_id_number":879336,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-17 15:58:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2023-04-12 12:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2023-04-17 08:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2023-04-17 13:43:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-03 12:58:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-03 14:15:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-11 15:32:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2023-04-17 08:32:13.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"c27e654d019640d5b3987f0173236fba5719279e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-03 12:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62fe5b913b769219c5b4f4383a159a29024dc8e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-03 12:33:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9325872ce60fd01b3aa8a60f24d9dd6456f40dd1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-03 12:38:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d755d7192652e1c92a08230abbbd05bd4d0e14b","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 12:58:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3caa59a4ef6c47039e16aa94c8318185\n\n- openstack-tox-py310-with-sqlalchemy-master-sfc https://zuul.opendev.org/t/openstack/build/ed1eb85555a546b99897ff8eb69b17fc : SUCCESS in 13m 31s\n- openstack-tox-py310-with-oslo-master https://zuul.opendev.org/t/openstack/build/d524ff62df4941aa84ed19a022c6f380 : SUCCESS in 12m 18s","accounts_in_message":[],"_revision_number":2},{"id":"159488867581fc340235cb811fe00462ed822c72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 14:15:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/868e05799a61469998e4cfc4f6569437\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8226e1e7310a4e8b8488b9936175476f : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/603f5f28d6e84013b3d4d711c350ef35 : SUCCESS in 6m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2765c2fd28034c949ac1699463535b75 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eda3bb1b6e14056a6ea961a7643d304 : SUCCESS in 9m 22s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/ebdb9a6ca94e41a48840efaecf4d7c35 : SUCCESS in 16m 27s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/d95d61f5b61d450fa511e766144c08c0 : SUCCESS in 1h 34m 43s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/5c3c3af01c344ee2a73ab49ac849f265 : SUCCESS in 1h 39m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3f5e85dbb6d4944529ed08804f83b81867bfb281","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-11 15:32:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50dcfbff45614777fd1eb142712cb43a8fddc8af","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-11 15:33:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c51802d0fc5d241723fd615a92923ac283defab4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-12 12:24:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c7c09f341e190f6d00576008da88d914c69b68c4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-12 12:57:52.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e7a7e7b7e9a34ad84b3c1e3c224248da98925619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 14:25:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a66f7ff568944ed80e2abd7febe84e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2edba05c16f42b38a31eb785a446381 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e9c7986c7604c14903199efb7cecdd7 : SUCCESS in 7m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/257ffb55e68e44f49b51aab18cd802fd : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47cac94ece744d8cacd4af7e1e2b37f0 : SUCCESS in 7m 59s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/ff994e7eb87c4fae9d60218b1a7ece3f : SUCCESS in 23m 05s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/c722d8cfa8bc41b1905bc4ee32fd168f : SUCCESS in 1h 22m 10s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/4fcd4ee53e9e4efab2292fe8ad9cf58d : SUCCESS in 1h 54m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e7fda5f6a4c631fab70891b9b7bcb4d3f5d8bd75","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-04-17 08:32:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"de0de87793f114307a4e3570db3c20827a0fa0b0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-04-17 13:43:49.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3ca51e31787e19c74f148e8b3091c5feb697f837","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:44:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7276ac01f4a8fedfa02b472da3655e30acfeede0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:58:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/954914a098934ccb83bd65f7483bb7f2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a121697176bb45418f92e413325c12b8 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b106f825489e4dbb81184741406a3516 : SUCCESS in 6m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f384c7a2c0054b9792423d29292f2fb8 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48cf19d9e50540aa88c14c5bb1ee5e80 : SUCCESS in 14m 47s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/0a4e6b1b2b9d47688ef027ffe962eaa9 : SUCCESS in 40m 23s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/9c24d6144d6a42078f4433c45df71bb7 : SUCCESS in 2h 05m 39s","accounts_in_message":[],"_revision_number":3},{"id":"513c20937d42bb90ed35c818cc39f11bbe2f0fcb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:58:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"faed7a097b3f5252d5f80ecc5ee98f9545b00af8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 16:01:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c63565494712464e944a3fa83a4bf6e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/317ec5f9112f46f6b3a50cac32bf54f5 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"887bed68c7ab52eacf1da99cafea17059a232ac1","revisions":{"d4b77393e2b7f84779fb7c9cb98f3f00b0492caa":{"kind":"REWORK","_number":1,"created":"2023-04-03 12:27:46.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/36/879336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/36/879336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/1"}}},"commit":{"parents":[{"commit":"f8e7ff9a21ddc48cfd859430ca6f1c3ba56287b3","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f8e7ff9a21ddc48cfd859430ca6f1c3ba56287b3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-03 12:26:33.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-03 12:26:33.000000000","tz":120},"subject":"Add neutron and neutron-lib projects to SQLAlchemy main branch job","message":"Add neutron and neutron-lib projects to SQLAlchemy main branch job\n\nChange-Id: I16a823d9274f4ec5290c1e2d5d009d02f35623c0\nRelated-Bug: #2004265\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d4b77393e2b7f84779fb7c9cb98f3f00b0492caa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d4b77393e2b7f84779fb7c9cb98f3f00b0492caa"}]},"branch":"refs/heads/master"},"3b6c5f67110bc0569baffd39a3574941634ed27a":{"kind":"REWORK","_number":2,"created":"2023-04-03 12:33:39.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/36/879336/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/36/879336/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/2"}}},"commit":{"parents":[{"commit":"f8e7ff9a21ddc48cfd859430ca6f1c3ba56287b3","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f8e7ff9a21ddc48cfd859430ca6f1c3ba56287b3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-03 12:26:33.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-03 12:33:26.000000000","tz":120},"subject":"Add neutron and neutron-lib projects to SQLAlchemy main branch job","message":"Add neutron and neutron-lib projects to SQLAlchemy main branch job\n\nChange-Id: I16a823d9274f4ec5290c1e2d5d009d02f35623c0\nRelated-Bug: #2004265\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3b6c5f67110bc0569baffd39a3574941634ed27a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3b6c5f67110bc0569baffd39a3574941634ed27a"}]},"branch":"refs/heads/master"},"887bed68c7ab52eacf1da99cafea17059a232ac1":{"kind":"REWORK","_number":3,"created":"2023-04-12 12:24:55.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/36/879336/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/36/879336/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/36/879336/3"}}},"commit":{"parents":[{"commit":"f8e7ff9a21ddc48cfd859430ca6f1c3ba56287b3","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f8e7ff9a21ddc48cfd859430ca6f1c3ba56287b3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-03 12:26:33.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-12 12:24:36.000000000","tz":120},"subject":"Add neutron and neutron-lib projects to SQLAlchemy main branch job","message":"Add neutron and neutron-lib projects to SQLAlchemy main branch job\n\nChange-Id: I16a823d9274f4ec5290c1e2d5d009d02f35623c0\nRelated-Bug: #2004265\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/887bed68c7ab52eacf1da99cafea17059a232ac1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/887bed68c7ab52eacf1da99cafea17059a232ac1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
