)]}'
{"id":"openstack%2Fnetworking-sfc~881348","triplet_id":"openstack%2Fnetworking-sfc~master~I079ce4192baf91b10c4a29a8344fd3feec055b44","project":"openstack/networking-sfc","branch":"master","topic":"bug/2017478","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-08-31 08:04:13.000000000","reason":"Change was abandoned"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2023-08-31 08:04:13.000000000","reason":"Change was abandoned"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2023-08-31 08:04:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I079ce4192baf91b10c4a29a8344fd3feec055b44","subject":"Move to python3.9 as minimal python version","status":"ABANDONED","created":"2023-04-24 10:00:01.000000000","updated":"2023-08-31 08:04:13.000000000","total_comment_count":5,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"66f00bfd06b8c6533cd96713f6e988ec8f8d7bdd","_number":881348,"virtual_id_number":881348,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-27 10:13:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":-1,"date":"2023-05-03 20:07:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-04-24 10:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-24 19:29:28.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2023-04-25 04:53:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2023-05-03 20:07:25.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"3c4c3c9376a710d9eb00361de8e7dff545116b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-24 10:00:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"093494be93d55a8fbfa2411285cfbbcd8b9b461e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 10:13:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eabe6c961c8d454488c029740e936670\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4e19ac3d2a76423299bd6bb1c5ba9785 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbbefdc3ba9f40a9b42fd94ab749618d : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5da27a4530624402b8d51245dbf14365 : SUCCESS in 7m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2efaff9a69ca4442b0a7ebbc3fc25743 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a937e62c8681437582d9cf87b829e9d8 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"66b5fad76647b9ef706968e3ca2f35b8783901d9","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2023-04-24 19:29:28.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9f1736e7c1e21830a7b98c1d7031f12142ac585","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-04-25 04:53:26.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cd360e88b8017d96af72dee665ea8d75cf35215","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-26 07:56:02.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23014e5d7f65b288e0893e522a4e48bd9b00f2c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-27 08:47:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2751a83ec47c274636421188e2b3a8b747532d36","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-04-27 08:48:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e0067082baac6857fc62506bfa120b586f7f70a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 10:13:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/334c45a243b84332a0475a0b4cf51aba\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3e4da51a3c47499bb0c268bcc3abab4a : SUCCESS in 1m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cccc06bdb85a4d1a9a932a544a00ca7e : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bbda3b20e044d348b4087b1f3f1706e : SUCCESS in 8m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7847e2439a24cd187ab2f9ba2e53c69 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696e30a9cf474ae8a370168b3e412353 : SUCCESS in 8m 53s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/d50e589e14524b01ae38b21ac8757e11 : SUCCESS in 21m 17s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/be5f95aa12ea4379a06b730b80d51ced : SUCCESS in 1h 24m 03s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/50cb2b6e9f79457285e58c172d323d9b : FAILURE in 29m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd3f6122ddf349d74c9133dd9bb323a2af1da6b1","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 10:21:00.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac563f4230f64ddfb945340118aeee55\n\n- openstack-tox-py310-with-sqlalchemy-main-sfc https://zuul.opendev.org/t/openstack/build/fee2a12bf06e42178cd5a82309ab84c3 : SUCCESS in 14m 08s\n- openstack-tox-py310-with-oslo-master https://zuul.opendev.org/t/openstack/build/7e037c86235d495ba24dad90f17f9a81 : SUCCESS in 11m 55s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/2f2818c891804124bb50ca84105d4fdb : SUCCESS in 1h 30m 53s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/031eb48d05f4431ba8876789f213dcbb : FAILURE in 16m 29s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/a633cfcce0e84157a640c3e246d81e96 : SUCCESS in 17m 40s","accounts_in_message":[],"_revision_number":2},{"id":"cb6ef965cc7829474e4ac85f896d4a2c2a334e5a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-03 20:07:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"66f00bfd06b8c6533cd96713f6e988ec8f8d7bdd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2023-08-31 08:04:13.000000000","message":"Abandoned\n\nIf we need the new python runtime we can reopen or re-propose these patches","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0fe826ff4d9c1cb4eb4bb356fdd6eeffbb1bd87d","revisions":{"75b6b0223c96e2b2089efbab66489bf1d8578f37":{"kind":"REWORK","_number":1,"created":"2023-04-24 10:00:01.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/881348/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/48/881348/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/881348/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/881348/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/881348/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/48/881348/1"}}},"commit":{"parents":[{"commit":"da019219b4b96f5d9f2317c6a7fe6eec61c47349","subject":"Merge \"Add neutron and neutron-lib projects to SQLAlchemy main branch job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/da019219b4b96f5d9f2317c6a7fe6eec61c47349"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-24 09:59:25.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-24 09:59:25.000000000","tz":120},"subject":"Move to python3.9 as minimal python version","message":"Move to python3.9 as minimal python version\n\n2023.2 release is targetting python3.9 as\nminimal testing rutime[1]. openstack-tox-py38\njob started to fail as libraries(like tooz) already\ndropped py3.8 support.\n\nThis patch updates networking-sfc jobs and\nsetup.cfg to reflect the same.\n\n[1] https://review.opendev.org/c/openstack/governance/+/872232\n\nChange-Id: I079ce4192baf91b10c4a29a8344fd3feec055b44\nCloses-Bug: #2017478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/75b6b0223c96e2b2089efbab66489bf1d8578f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/75b6b0223c96e2b2089efbab66489bf1d8578f37"}]},"branch":"refs/heads/master"},"0fe826ff4d9c1cb4eb4bb356fdd6eeffbb1bd87d":{"kind":"REWORK","_number":2,"created":"2023-04-27 08:47:27.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/48/881348/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/48/881348/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/881348/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/881348/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/48/881348/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/48/881348/2"}}},"commit":{"parents":[{"commit":"da019219b4b96f5d9f2317c6a7fe6eec61c47349","subject":"Merge \"Add neutron and neutron-lib projects to SQLAlchemy main branch job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/da019219b4b96f5d9f2317c6a7fe6eec61c47349"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-24 09:59:25.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2023-04-27 08:47:04.000000000","tz":120},"subject":"Move to python3.9 as minimal python version","message":"Move to python3.9 as minimal python version\n\n2023.2 release is targetting python3.9 as\nminimal testing rutime[1]. openstack-tox-py38\njob started to fail as libraries(like tooz) already\ndropped py3.8 support.\n\nThis patch updates networking-sfc jobs and\nsetup.cfg to reflect the same.\n\n[1] https://review.opendev.org/c/openstack/governance/+/872232\n\nChange-Id: I079ce4192baf91b10c4a29a8344fd3feec055b44\nCloses-Bug: #2017478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0fe826ff4d9c1cb4eb4bb356fdd6eeffbb1bd87d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/0fe826ff4d9c1cb4eb4bb356fdd6eeffbb1bd87d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
