)]}'
{"id":"openstack%2Fnetworking-sfc~904208","triplet_id":"openstack%2Fnetworking-sfc~master~Idb82fbd93eff2e5ff4873ecd42af471692c850a4","project":"openstack/networking-sfc","branch":"master","topic":"python-3.12","attention_set":{},"removed_from_attention_set":{"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2024-04-04 09:01:58.000000000","reason":"Change was abandoned"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2024-04-04 09:01:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idb82fbd93eff2e5ff4873ecd42af471692c850a4","subject":"Python 3.12: assertDictContainsSubset is removed","status":"ABANDONED","created":"2023-12-21 13:31:15.000000000","updated":"2024-04-04 09:01:58.000000000","total_comment_count":12,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"d05063caa663a9d4f07c19aeff911969470c761a","_number":904208,"virtual_id_number":904208,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-01-09 12:48:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":-1,"date":"2024-02-01 10:10:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-21 14:49:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-03 12:32:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-01-10 12:36:23.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-01-23 08:37:08.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-02-01 10:10:31.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-04-04 09:01:58.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"d91e1a46c436a2758929fd6c88dec5e833b4b732","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2023-12-21 13:31:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59e03bdd8fea82810c0648bb4dfdee84dd763c5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 14:49:49.000000000","message":"Patch Set 1: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8168bd82d6f4106becee8913b13732f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4709d70ed68a4f49b79f1ff8a82a1be5 : FAILURE in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87f0efd52d0b41d9a4c7c96426c0f5a7 : FAILURE in 9m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a36768abc24417ab2167bace0f38adb : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce3266842bbe4951955bd308f146fc25 : SUCCESS in 9m 56s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/0b6773fe5ed44ff59170ce29b0b320d4 : SUCCESS in 17m 11s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/6e1e4b2a39e64495995e686c4ded6a53 : SUCCESS in 1h 15m 20s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/64dd6e3f9a4f40cd9bf5c02b8c2a34f2 : FAILURE in 11m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"077c5484e50fcdff99fec29628cc36d952e71bdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-01-09 10:48:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"31995ea44ce08905caed089a08935038af385d84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 12:48:53.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e998a1f4be954e568f7e0935fad1e1ac\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417493c47430485188fad8ee62e354f8 : FAILURE in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd5c9da1aad34a64b5260d57584dacd8 : FAILURE in 6m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68b7efc76e744cff91159337ab64f34d : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ae3e100e19445dbccb5709620ef64a : SUCCESS in 8m 38s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/90b1b10a7ff948b892149d1c5e5b7ee9 : SUCCESS in 13m 39s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/987dddf1fbec4746bf7700429b5b1303 : SUCCESS in 1h 39m 12s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/90d6c0e78fe7493a8b5703e94290466c : FAILURE in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"036251d9910f14ec14ceecf8f3f8178a857be4fe","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-10 12:36:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6fa8000be14eda84b24fa1b882fce8525f4ac13b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-11 09:40:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91954c0f4f04f1cba40ada81cf63d79b8bb738b5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-11 10:09:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a019d865bf1144e50c1489e88d284a3cc33eb7d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-11 10:10:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c53dd2a24fd58128ea0ac7c9bad31ec293c3a385","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-11 10:11:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b1fce9bad538575b032426d8fdaf57c5d9cd14b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-01-15 09:03:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ef848d2e957289a96c880b09e6d324bc89815696","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 08:37:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"016b2f415fe98869c2e46d574b51fa42c585cf98","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-31 17:05:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9bbb729ed729a403ded483292188bfc90b5ae040","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-31 17:07:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b2c932bac43602d643065d2eb79391d81b541cd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-31 17:13:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3bd09a60a35d97f56097690696d6c0f299398db","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-01 10:10:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d05063caa663a9d4f07c19aeff911969470c761a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-04-04 09:01:58.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd949b6940a17415b504ea67dac3f21129bd36bf","revisions":{"eb210651df0d8f2893f1fcecfd703b6f1c63b331":{"kind":"REWORK","_number":1,"created":"2023-12-21 13:31:15.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/08/904208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/08/904208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/08/904208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/08/904208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/08/904208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/08/904208/1"}}},"commit":{"parents":[{"commit":"cdde6a9ba20e500e55ffd6c1f73eb5c38eb1e8b9","subject":"Remove usage of LBaaS constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/cdde6a9ba20e500e55ffd6c1f73eb5c38eb1e8b9"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-12-21 13:30:09.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-12-21 13:31:13.000000000","tz":60},"subject":"Python 3.12: assertDictContainsSubset is removed","message":"Python 3.12: assertDictContainsSubset is removed\n\nassertDictContainsSubset(a, b) is removed from Python 3.12,\ninstead, we can use assertEqual(b, b | a), which is the\nrecommended way to do things. This patch implements it.\n\nChange-Id: Idb82fbd93eff2e5ff4873ecd42af471692c850a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/eb210651df0d8f2893f1fcecfd703b6f1c63b331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/eb210651df0d8f2893f1fcecfd703b6f1c63b331"}]},"branch":"refs/heads/master"},"cd949b6940a17415b504ea67dac3f21129bd36bf":{"kind":"REWORK","_number":2,"created":"2024-01-09 10:48:54.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/08/904208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/08/904208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/08/904208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/08/904208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/08/904208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/08/904208/2"}}},"commit":{"parents":[{"commit":"cdde6a9ba20e500e55ffd6c1f73eb5c38eb1e8b9","subject":"Remove usage of LBaaS constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/cdde6a9ba20e500e55ffd6c1f73eb5c38eb1e8b9"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2023-12-21 13:30:09.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-01-09 10:48:46.000000000","tz":60},"subject":"Python 3.12: assertDictContainsSubset is removed","message":"Python 3.12: assertDictContainsSubset is removed\n\nassertDictContainsSubset(a, b) is removed from Python 3.12,\ninstead, we can use assertEqual(b, b | a), which is the\nrecommended way to do things. This patch implements it.\n\nChange-Id: Idb82fbd93eff2e5ff4873ecd42af471692c850a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/cd949b6940a17415b504ea67dac3f21129bd36bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/cd949b6940a17415b504ea67dac3f21129bd36bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
