)]}'
{"id":"openstack%2Fnetworking-sfc~907531","triplet_id":"openstack%2Fnetworking-sfc~master~Ib52e4ce1e9e48b40774a29711b1cc9d20265d07e","project":"openstack/networking-sfc","branch":"master","topic":"bump-hacking","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-02-14 10:23:05.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-02-18 18:28:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib52e4ce1e9e48b40774a29711b1cc9d20265d07e","subject":"Bump hacking","status":"MERGED","created":"2024-02-02 11:25:54.000000000","updated":"2024-02-18 18:29:39.000000000","submitted":"2024-02-18 18:28:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"907531-bump-hacking","meta_rev_id":"06057f528506d6ff4c397c623b21877514db94f4","_number":907531,"virtual_id_number":907531,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-18 18:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-02-14 10:23:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-02-18 16:53:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-02 13:15:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-09 10:46:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"83a3830118d8c9b026b2f6f4409243b103489353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-02 11:25:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1dfb075e6aeca2542ca21365cb42f13bf95248f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-02 13:15:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c59baab42a34989a4a28edfd89e63ec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201b221ac5954e54b0591a538b8ebc09 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e3f9923768348b391833127177d4ad1 : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33485c9fd41948989bbb1f8aeb9c58c3 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48531842d2b4695ba6da46e4b7faf76 : SUCCESS in 8m 03s\n- requirements-check https://zuul.opendev.org/t/openstack/build/34c7247c90424a8ba5823eddf974bac4 : SUCCESS in 2m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2536238dafac4dca873c82dbb8b7dee1 : SUCCESS in 4m 00s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/89234a3b38574034ba753517cea7051c : SUCCESS in 30m 56s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/b003fdf1a89f4a2db9346a98e06b3e36 : SUCCESS in 1h 42m 06s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/fe031afcb1bc46e298c13fa3cf7a0c6e : SUCCESS in 1h 48m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d86e5b6a902c876a28843b5532eb9e34118f3a09","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-06 13:09:55.000000000","message":"Topic dontdelete/bump-hacking removed","accounts_in_message":[],"_revision_number":1},{"id":"a999e88d5e56f3e2143a5ccaa16a3fa12ee356ca","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-06 13:10:03.000000000","message":"Topic set to bump-hacking","accounts_in_message":[],"_revision_number":1},{"id":"c02fad63ae50c11a018e989f21c7c57cf37688a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-09 09:48:21.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2d7c2e77ef17a1ed260cb862800665741c81b966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 11:38:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3fb115ab11a4513b3a1697e4dea6243\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86deb86283b8406c88cd513aec7dba48 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd3998080d7f4e4bbaba0c052a5365f6 : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a234ce4fd139451eadcb00c21a2a0cfd : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017a595bd350459e97146e741c231f82 : SUCCESS in 9m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/529b6b9db6fb47e7a11633d7ef10b4fb : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed71b6eba04c42d79b8b2553c516d247 : SUCCESS in 2m 31s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/c4e9514dc4ff4a3cad6a2474977d2224 : SUCCESS in 16m 19s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/b2a6c1973ad74ab3b368e4cca652c670 : SUCCESS in 1h 44m 28s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/b56367789aee49f7adca9c9e190080fa : SUCCESS in 1h 36m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8d7c2988ae8c44cf7a3f462e173792534a5aad9a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-12 17:41:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1363792dd283261e7d4938f207b0a254311e492c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-14 10:09:15.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a56c21d70ed14efba8e7734191503ade12023ae0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-14 10:23:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3b14297185c199c9e1d9ce3406843ace74c67f16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 11:47:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9aba6e8c40944e5b521286bce1afa93\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdaa97c2da3e4816b4042570e9a22e21 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6235b2f15d884f58b7a8be5bc8c7aa9d : SUCCESS in 7m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ab3d4bcd04d44e8aa1756347b5b9e9d : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/058faf9090ce47768029d034ebd7518d : SUCCESS in 8m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2fbad04e0a9e4a018a90e041d913b4a1 : SUCCESS in 2m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbf3c95426a14246ac17ec89c81e0d01 : SUCCESS in 2m 41s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/fa6166feafdf41b4b4f1dc0a2fb90be1 : SUCCESS in 23m 49s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/500a2c863d5241b78cd9ba4ae135fef6 : SUCCESS in 1h 18m 31s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/3c3c98b5af1347e6a1e4fca87a779544 : SUCCESS in 1h 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f93c802f3c88f5da7d525dcf43d2d22170a441e6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-18 16:53:36.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a2bd0524a0698a90afca7eae47f79823bb39906d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 16:54:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9b2b158ac02081d10fa3154079bacff2565accc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 18:28:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0e6cedb472a44a085ffbc149a072fd3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c36519119e4b44b9aef887f37c58dfed : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/819593139cc24bfca19be72d703f687f : SUCCESS in 7m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd8b38ef590a43afb87044e2cd3c6589 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71a4de66588b4ccc9c6f9209b8a167c8 : SUCCESS in 8m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/127f4db2a0f14410a1aab33d90d21ae1 : SUCCESS in 1m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/274dad9e912649458d64fe42ffc38fa8 : SUCCESS in 2m 37s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/0b5d35cfe43d4fdb8c74d13570b85bb7 : SUCCESS in 24m 10s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/f94d011429994de48c79a81d04bfc352 : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":3},{"id":"9c8ed94b95c44ec00bc10ccfc531f13995290863","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 18:28:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"06057f528506d6ff4c397c623b21877514db94f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 18:29:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc4c283e4c7b481aac6168e66cc0f957\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18208150ba9a4c4db4da93c61bd81b01 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b987e10ec7004af9aa9deaf964258f65 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"54834f192dde52bb8e67908885195bb0d308dc52","revisions":{"7e59af3493fb518172c200a412fd1f73b151b8a8":{"kind":"REWORK","_number":1,"created":"2024-02-02 11:25:54.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/31/907531/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/31/907531/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/1"}}},"commit":{"parents":[{"commit":"6bc53e15d3f2c4bfd0eba01e632a5d7cbe0b88f9","subject":"Merge \"Update python classifier with py3.10 \u0026 py3.11 in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6bc53e15d3f2c4bfd0eba01e632a5d7cbe0b88f9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-02-02 11:25:33.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-02-02 11:25:33.000000000","tz":60},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old.\nTry to synchronize pylint ignore and extension list with\nother Networking projects.\n\nChange-Id: Ib52e4ce1e9e48b40774a29711b1cc9d20265d07e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7e59af3493fb518172c200a412fd1f73b151b8a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/7e59af3493fb518172c200a412fd1f73b151b8a8"}]},"branch":"refs/heads/master"},"3f56ce8733597ae14420eb5903d4ad510dddc80e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-09 09:48:21.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/31/907531/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/31/907531/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/2"}}},"commit":{"parents":[{"commit":"6bc53e15d3f2c4bfd0eba01e632a5d7cbe0b88f9","subject":"Merge \"Update python classifier with py3.10 \u0026 py3.11 in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6bc53e15d3f2c4bfd0eba01e632a5d7cbe0b88f9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-02-02 11:25:33.000000000","tz":60},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2024-02-09 09:48:21.000000000","tz":0},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old.\nTry to synchronize pylint ignore and extension list with\nother Networking projects.\n\nWith new pip the order of packages is not relevant, so the\nrelated comment from requirements.txts is removed, see pip\ndocumentation:\nhttps://pip.pypa.io/en/stable/cli/pip_install/#installation-order\n\nChange-Id: Ib52e4ce1e9e48b40774a29711b1cc9d20265d07e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3f56ce8733597ae14420eb5903d4ad510dddc80e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3f56ce8733597ae14420eb5903d4ad510dddc80e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"54834f192dde52bb8e67908885195bb0d308dc52":{"kind":"REWORK","_number":3,"created":"2024-02-14 10:09:15.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/31/907531/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/31/907531/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/31/907531/3"}}},"commit":{"parents":[{"commit":"6bc53e15d3f2c4bfd0eba01e632a5d7cbe0b88f9","subject":"Merge \"Update python classifier with py3.10 \u0026 py3.11 in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6bc53e15d3f2c4bfd0eba01e632a5d7cbe0b88f9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-02-02 11:25:33.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-02-14 10:09:05.000000000","tz":60},"subject":"Bump hacking","message":"Bump hacking\n\nhacking 3.0.x is too old.\nTry to synchronize pylint ignore and extension list with\nother Networking projects.\n\nWith new pip the order of packages is not relevant, so the\nrelated comment from requirements.txts is removed, see pip\ndocumentation:\nhttps://pip.pypa.io/en/stable/cli/pip_install/#installation-order\n\nChange-Id: Ib52e4ce1e9e48b40774a29711b1cc9d20265d07e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/54834f192dde52bb8e67908885195bb0d308dc52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/54834f192dde52bb8e67908885195bb0d308dc52"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
