)]}'
{"id":"openstack%2Fnetworking-sfc~907984","triplet_id":"openstack%2Fnetworking-sfc~master~Ib1b906805a7e099e3d14b7ad2491ccdd1e6f0ed1","project":"openstack/networking-sfc","branch":"master","topic":"tox-envdir","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-02-13 20:37:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib1b906805a7e099e3d14b7ad2491ccdd1e6f0ed1","subject":"tox: Drop envdir","status":"MERGED","created":"2024-02-06 10:37:19.000000000","updated":"2024-02-13 20:38:55.000000000","submitted":"2024-02-13 20:37:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907984-tox-envdir","meta_rev_id":"d77711222c48e72932276584f2ce92ec8c22da0d","_number":907984,"virtual_id_number":907984,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-13 20:37:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-02-13 19:10:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2024-02-13 15:13:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-02-07 16:01:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-02-13 19:10:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 12:14:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 16:01:31.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-02-13 15:13:58.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-02-13 19:10:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"00064448eb7a3a22151c311cf3d09709ae8852da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-02-06 10:37:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8339af0808b1e6e70846b8a986330576b5bcf36b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 12:14:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c6b7da1122640f18b3ccb27a39ae350\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0fbfc245e4f43eeb6f958c5fee825ef : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/add8729a87cd49df9bf46c17060b4860 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1837286aca1b420badebe3fa44ccbee1 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d0229a372b44699b1acfb25da2c8002 : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/038c2a01d6894655b96528e903aaf157 : SUCCESS in 3m 50s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/9b229698495c4a59800285e1eea752df : SUCCESS in 15m 48s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/dfdd0023f76b4aed9d7549990794a262 : SUCCESS in 1h 31m 44s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/8358b89ddab34014b93876867a7ee950 : SUCCESS in 1h 32m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7a20ca2e39db906acfde68deaf59c2f0236248b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-02-07 16:01:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1e244ce43559af0bf6e08132788f7676688d9237","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-02-13 15:13:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bba6c8c4ac67cd5599316e5bc458b9d41711a71a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-02-13 19:10:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c25b743175971ff80f11dd404afd0bb677abf8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 19:10:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5437c0447a95ec1bdd0f2727b6d36ea7d217c2ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 20:37:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2327a08424ce45cf998140d6d9dc0e08\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/698ca718218d4c93a960653512a29f4a : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30fdf07208c94787ab3ba83adab3c067 : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9ad24751f9f453a82ec83e42a92382e : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70119979885242c49b70e05365a094ad : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0600db8c74634f05a85c0005da892b84 : SUCCESS in 2m 27s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/06db6ed01b4346b1a9ae13943460408e : SUCCESS in 16m 44s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/ff8060b5d0ee4bc6b48b17802bef3ce2 : SUCCESS in 1h 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"68109b851f3986c411c0bc49e8bb0aa3c7ef7df8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 20:37:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d77711222c48e72932276584f2ce92ec8c22da0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 20:38:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56b17d97c2a4ae389c58e488c60865b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b167e9d5acd048fcbd8d9974a473d8da : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36ee302a7d7c4bb1a1c3f87d231c6971 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbac81f212b8a45376b07e48c5c5b911a55a209b","revisions":{"cbac81f212b8a45376b07e48c5c5b911a55a209b":{"kind":"REWORK","_number":1,"created":"2024-02-06 10:37:19.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/84/907984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/84/907984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/84/907984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/84/907984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/84/907984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/84/907984/1"}}},"commit":{"parents":[{"commit":"6bc53e15d3f2c4bfd0eba01e632a5d7cbe0b88f9","subject":"Merge \"Update python classifier with py3.10 \u0026 py3.11 in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/6bc53e15d3f2c4bfd0eba01e632a5d7cbe0b88f9"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-02-06 10:36:48.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-02-06 10:36:48.000000000","tz":60},"subject":"tox: Drop envdir","message":"tox: Drop envdir\n\ntox now always recreates an env although the env is shared using envdir\noptions.\n~~~\n$ tox -e genpolicy\ngenpolicy: recreate env because env type changed from\n{\u0027name\u0027: \u0027genconfig\u0027, \u0027type\u0027: \u0027VirtualEnvRunner\u0027} to\n{\u0027name\u0027: \u0027genpolicy\u0027, \u0027type\u0027: \u0027VirtualEnvRunner\u0027}\n~~~\n\nAccording to the maintainer of tox, this functionality is not intended\nto be supported.\nhttps://github.com/tox-dev/tox/issues/425#issuecomment-1011944293\n\nChange-Id: Ib1b906805a7e099e3d14b7ad2491ccdd1e6f0ed1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/cbac81f212b8a45376b07e48c5c5b911a55a209b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/cbac81f212b8a45376b07e48c5c5b911a55a209b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
