)]}'
{"id":"openstack%2Fnetworking-sfc~917971","triplet_id":"openstack%2Fnetworking-sfc~master~I856627010be25490aacfa6e8fa9eef8b25b92554","project":"openstack/networking-sfc","branch":"master","topic":"pre-commit-hook","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-08-26 09:31:16.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-08-26 09:31:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I856627010be25490aacfa6e8fa9eef8b25b92554","subject":"Remove executable from python files which don\u0027t really needs it","status":"MERGED","created":"2024-05-02 14:30:42.000000000","updated":"2024-08-26 09:32:17.000000000","submitted":"2024-08-26 09:31:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917971-pre-commit-hook","meta_rev_id":"688794f520eddb16f148022e2841597ae5769eaa","_number":917971,"virtual_id_number":917971,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-26 09:31:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-08-26 07:33:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-08-24 17:16:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2024-08-24 17:16:34.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-08-26 07:33:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-02 16:21:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 14:57:23.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-05-16 16:00:04.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-08-26 07:33:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"18247f170a068c0bd60bfcca35da29cc813fd502","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-05-02 14:30:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ba5625bdc62938ca29fee566cea4e56d473ba0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 16:21:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c78a2f79aa6a42608758cee912551ea6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dd4072e469b4c499d814bf53396934a : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdede5114a2a45b28c5bbce87b85ab17 : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd4cb52f486b46a4ad9ae534e6f87478 : SUCCESS in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95733245701243abb64384ccf03aea23 : SUCCESS in 8m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0a15170db4c4d1db9caba705dcc31b6 : SUCCESS in 7m 44s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/4a4edbc98a9c458ab46e9fc30c14693f : SUCCESS in 15m 05s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/d7bad6dc5c484aa48e56044b5f0cf2bc : SUCCESS in 1h 17m 06s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/b2ff4a83c0a74213874e94b91c40c627 : SUCCESS in 1h 33m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee3d9e1874011538622db183944541a79a5c756a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-05-13 14:57:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aaae12b044e829276d158cde1ecc54d4a509929d","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-16 16:00:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"93113c71871c03589045283eb254b5729a148798","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-30 15:44:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4abbe42522b4c47d366b9d1513670974704e12e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 15:45:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f83b4855373558712a78c412c050e8ebe8d9ae4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 18:43:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bae922ae563403d9e62bb09d3f924b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c4db84ff664b8e8effc52c5a2c63cc : SUCCESS in 9m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67503ef518fc449a8887a6cb879fbb8d : SUCCESS in 7m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/33b349735086404da9145b3ddc8c9842 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10ac40fdc7e4b6d9ce4862c676b1d3d : SUCCESS in 7m 28s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/c63afdec2c9f44359128f80b8414a859 : SUCCESS in 14m 30s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/51e6ee0f8e154479ad3bef3e0eb42cd5 : FAILURE in 2h 51m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2f779308f942375c41fe770cedc7fd828dca8090","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-30 19:22:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"948595ec47c808e2e68b577b5ef4ad6ad707856e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 22:24:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/573be8aaee964bdcbdbafb0ec529df38\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ea3921e1bfc48feb36a10a02f30f88e : SUCCESS in 10m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3ebdb175b754367b191e8c78aaa1b74 : SUCCESS in 8m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9c58e00691654f29bd3572805b49c3df : SUCCESS in 7m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60567b5fa4e547da9bd3747e8ec4e356 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6e285997ef441687770c6ff3ba041f : SUCCESS in 7m 31s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/5834da1b29ec47b291c452fc7d78b950 : SUCCESS in 22m 25s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/7db818c3262c4531ab540c31e3e6d005 : FAILURE in 2h 38m 09s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/ec10d96203fa47bc98bdedf9dd3f3063 : FAILURE in 2h 46m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d359623abbee68a8e735ba92023a22b22cd695d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-19 11:05:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ffac138b54447f591dd1b385133b7edcb5515c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 14:01:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3fbf3da77404894ac60799f4c3cffb3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2380a6c6cbb7432e96ddf24d1caa1389 : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb5af7daf47c4b53ab769ce91cd05107 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9e59b81be9145e1b2b0b89cf6f344fc : SUCCESS in 5m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/773fa3d0882646aba956f51bd033e02e : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17581554eef7489caa124db45b841026 : SUCCESS in 7m 13s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/e3b44cf3688a4761b2b836983ee2e935 : SUCCESS in 12m 16s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/7f842c93eb674d4aaec70bc70536f3ac : FAILURE in 2h 42m 48s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/e3904e414c2344549c906dd0a84a4f94 : FAILURE in 2h 50m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0889fc84ec03b04a9823ea44ceef00a62c927d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-24 17:16:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ca32b828b5167e9d8fb58ecc134382aff1e97348","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 19:17:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2db68d6bd55d42dc819084201b358436\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9af1938960d40c48639da88a1bd922a : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4734a440d6e449c8935586b949dd33d4 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d2470340c8d45a5a91ec6fd0597a388 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fa309a7014e492080d13b88b8a78220 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a66adf8f0f46420eb35f53faafbe02f6 : SUCCESS in 6m 35s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/b3d15bf9775a4453a15825240d2deafd : SUCCESS in 23m 23s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/960abd06e8a242f3878de6c0dcf790a4 : SUCCESS in 1h 16m 01s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/712cc5c0670443baa9f8c57acf7ded0a : SUCCESS in 2h 00m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dcaa847317669070354a21cea9055bd229184a23","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-26 07:33:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"18a56d4e2d91f66e6b91acafe44667675c39d354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 07:33:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1ae6a6e12e5d8e5604cb11e9498d3dd9077d514d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 09:31:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe8a0e981bbf47e99599a4c126313137\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ec56c5229fa4608ba17259732b90c9c : SUCCESS in 8m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a04f00731f942b3aff7a9e364857d4c : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e559bfbf8d47485abee706c42fb6ed84 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63014dc522ba48609607a638efa75ddd : SUCCESS in 8m 29s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/0158b7a05e614a0d9a3bfac8066dca84 : SUCCESS in 23m 49s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/980321bbd7d0496d8fefd7591e68dc5e : SUCCESS in 1h 52m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d1d5ff0c8511810406955494f0e726906ceff364","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 09:31:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"688794f520eddb16f148022e2841597ae5769eaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 09:32:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09350256fed54564adf45e6fc21cd01f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f73a4e6b565743b9adee7b6e97c1b9c6 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4b86661ee3db4afdd8df6614b645ec5a138997e9","revisions":{"feb59e15c0d284ff273579f74fbd55ac9447488f":{"kind":"REWORK","_number":1,"created":"2024-05-02 14:30:42.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/71/917971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/71/917971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/71/917971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/71/917971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/71/917971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/71/917971/1"}}},"commit":{"parents":[{"commit":"bbd2dbcc4b7a9e738cccfd64abd9631e9f1b0194","subject":"Merge \"python 3.12: assertDictContainsSubset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bbd2dbcc4b7a9e738cccfd64abd9631e9f1b0194"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-05-02 14:25:56.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-05-02 14:29:53.000000000","tz":120},"subject":"Remove executable from python files which don\u0027t really needs it","message":"Remove executable from python files which don\u0027t really needs it\n\nTrivialFix\n\nChange-Id: I856627010be25490aacfa6e8fa9eef8b25b92554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/feb59e15c0d284ff273579f74fbd55ac9447488f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/feb59e15c0d284ff273579f74fbd55ac9447488f"}]},"branch":"refs/heads/master"},"4b86661ee3db4afdd8df6614b645ec5a138997e9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-24 17:16:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/71/917971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/71/917971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/71/917971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/71/917971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/71/917971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/71/917971/2"}}},"commit":{"parents":[{"commit":"94d303464bfc9dc2eb244004924ad685e78053f8","subject":"Fix _delete_ports_flowrules_by_id so that it expects only one port_id","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/94d303464bfc9dc2eb244004924ad685e78053f8"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-05-02 14:25:56.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-08-24 17:16:34.000000000","tz":0},"subject":"Remove executable from python files which don\u0027t really needs it","message":"Remove executable from python files which don\u0027t really needs it\n\nTrivialFix\n\nChange-Id: I856627010be25490aacfa6e8fa9eef8b25b92554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4b86661ee3db4afdd8df6614b645ec5a138997e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4b86661ee3db4afdd8df6614b645ec5a138997e9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
