)]}'
{"id":"openstack%2Fnetworking-sfc~917972","triplet_id":"openstack%2Fnetworking-sfc~master~I9801bc7f23add5d4540913ce12eab1716dec62e5","project":"openstack/networking-sfc","branch":"master","topic":"pre-commit-hook","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-08-26 09:31:17.000000000","reason":"Change was submitted"},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2024-08-26 09:31:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9801bc7f23add5d4540913ce12eab1716dec62e5","subject":"Add pre-commit configuration","status":"MERGED","created":"2024-05-02 14:30:42.000000000","updated":"2024-08-26 09:33:06.000000000","submitted":"2024-08-26 09:31:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917972-pre-commit-hook","meta_rev_id":"52a0245ac1928292638485c44e6522ddd9792651","_number":917972,"virtual_id_number":917972,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-26 09:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-08-26 07:33:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2024-08-24 17:16:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2024-08-24 17:16:44.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-08-26 07:33:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-02 16:22:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 14:58:08.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2024-05-16 16:00:26.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2024-08-26 07:33:17.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"18d04af2b683ef6c56ac4687b8db415ac0c37b16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-05-02 14:30:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61efee578c11c9868a0670784b03b033c7511f23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 16:22:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe9e0acc4134c968361b7c4447d58a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b448bf5b10b14683b0826d1be3993fb0 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3f527ca3b8440ae854cab607ef00157 : SUCCESS in 7m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9dda588a15c84b1b9f63c9ff958b473e : SUCCESS in 8m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/055791c5ebeb469ca2a785c9cda043bd : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fde54920b030460cb7307d30592a004f : SUCCESS in 7m 11s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/a564ee2681644c1d9faa5a31bb69a695 : SUCCESS in 26m 21s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/c501f12a02a3403f9c6db1f4f4accdbc : SUCCESS in 1h 45m 22s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/aa4b9ccb7b6a42019adca6cf420b0bbf : SUCCESS in 1h 30m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70cae3b190aac5963ad231ccbedb0094e09c1fe8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2024-05-13 14:58:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7de4afc89b7f51a450247e8bf8759e20d4b8560b","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-16 16:00:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dbe17fb6c8188cfd626f2a0f01c693490803d518","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-30 15:45:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f28a4e225701e0f521acfe2c725efdd0e7c9605a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 15:45:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2de4e7402085e2da08b617a7af833320ff46e930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 18:43:28.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/917971 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"23c45ad62a43b3f04297c369012a960dd57ddb80","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-05-30 19:22:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d54e9c4a7f4ce7fa706d00ff0ec2d310ee53c47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 22:19:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7df47a513b1148fab01935330c019f7b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d74e91d96ec47d48d5c55dc1ddbb57b : SUCCESS in 9m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6d54f057ad74384977ecc0199589271 : SUCCESS in 9m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77fbbc7fef124ec9813fb659ed5eadc6 : SUCCESS in 9m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37cc72343d264e14a5e95e9b01ab67da : SUCCESS in 14m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9689dce4156a49f49aa93103a4aeb7e1 : SUCCESS in 7m 07s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/bd244ce3741a4f5dba176bd5e18d8436 : SUCCESS in 17m 02s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/00b87493987c47aea384fb65c7938c90 : FAILURE in 2h 47m 14s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/42be5712affe435d9377250f16547510 : FAILURE in 2h 51m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8baf086d3898677653ea77d830432ccd8d4fcb85","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-07-19 11:05:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b04534116b82a864391bb813fd3ab94f6ed3fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 13:57:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82e6b36052dc4e06bbfc5f6c63ffe48f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6ae54190bcf4472aaa5d91617c5039f : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/584dec27fb9d4689b06816101bc2cc6f : SUCCESS in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/72904580a4f843aea26905a47de67c27 : SUCCESS in 6m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9d9b8c3486042edbe4f1d20b7c3c90d : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d85ff400e316442987acc670e1090afe : SUCCESS in 7m 39s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/e51ccf00978a4ef28830791ed3f42489 : SUCCESS in 11m 19s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/a72e7afd39cd4477add6c5d02400d1ba : FAILURE in 2h 46m 35s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/f403e38e578d46239e69d1f3189b9a98 : FAILURE in 2h 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4996b2454a41179d3321170f818e65519bb57ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-24 17:16:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"49c5aeb0efe35abba6170967cb6f98e258fd0ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 19:00:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73717a06bcea48a8b3279040deeef877\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15f6c54d7f7e46cfbe4efe23755c34c6 : SUCCESS in 8m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d149169f727f42acab8acc4fee3dc1c1 : SUCCESS in 6m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b83254036142400c855420cad29ca814 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b64d0c9ca19a4787a6d696645e4f61b1 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/929de421825e4637b09253cd778b554c : SUCCESS in 6m 36s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/273f3358299f4167ae1e5b6f3f74c6b6 : SUCCESS in 22m 23s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/b1410d88c58b4fb1a89c1f14195cd6c4 : SUCCESS in 1h 43m 01s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/b3ba23d437394ffcb8a0f5122b6bd2b7 : SUCCESS in 1h 39m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"415f9db74ccf37b54aac882449a2b188744a93c6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-08-26 07:33:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"de37e7ba1276e902b024b6037da5a4e8fa210877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 07:33:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"208ca9a88f7679f26535a71eb257b3fa2217e075","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 09:31:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7a2b9e84485431ebef6c433da4fb69e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2657188f7b4a4aafbaf23da302739707 : SUCCESS in 10m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/911d0a776aa24bc297357cf1c30d8612 : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0095d8572e0e48e2b4a7cca30cd62af8 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75b778b36e4240859d6f6d43890ed524 : SUCCESS in 7m 14s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/9c37717dd32a44628bec6212553b3c61 : SUCCESS in 24m 56s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/26c724740f2242298511af0a81475b27 : SUCCESS in 1h 37m 18s","accounts_in_message":[],"_revision_number":2},{"id":"59fed5cc5c31f657eaff8c61c7b93c4013984939","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 09:31:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"52a0245ac1928292638485c44e6522ddd9792651","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 09:33:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9445830d2844cab81a97e16ad05dd8d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d49fadb777e1415ea28b4a1fdc2bee78 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fb51aa5ccbd279f69dfa978e2a4dcdffece865e9","revisions":{"58725de4c31f928cfa067b9a50159a95da34aff4":{"kind":"REWORK","_number":1,"created":"2024-05-02 14:30:42.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/72/917972/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/72/917972/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/72/917972/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/72/917972/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/72/917972/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/72/917972/1"}}},"commit":{"parents":[{"commit":"feb59e15c0d284ff273579f74fbd55ac9447488f","subject":"Remove executable from python files which don\u0027t really needs it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/feb59e15c0d284ff273579f74fbd55ac9447488f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-05-02 14:30:20.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-05-02 14:30:20.000000000","tz":120},"subject":"Add pre-commit configuration","message":"Add pre-commit configuration\n\n\u0027pre-commit\u0027 tool [1] provides an extensible way to configure hooks\nwhich can be run automatically before each commit. That way we can make\nsure that code send to review is passing base static analysis checks\nlike e.g. bandit, flake8 or hacking.\n\nFor now this is optional to use and not integrated in e.g. tox.ini file\nbut we can integrate it there too as next step.\n\n[1] https://pre-commit.com/\n\nChange-Id: I9801bc7f23add5d4540913ce12eab1716dec62e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/58725de4c31f928cfa067b9a50159a95da34aff4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/58725de4c31f928cfa067b9a50159a95da34aff4"}]},"branch":"refs/heads/master"},"fb51aa5ccbd279f69dfa978e2a4dcdffece865e9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-24 17:16:44.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/72/917972/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/72/917972/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/72/917972/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/72/917972/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/72/917972/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/72/917972/2"}}},"commit":{"parents":[{"commit":"4b86661ee3db4afdd8df6614b645ec5a138997e9","subject":"Remove executable from python files which don\u0027t really needs it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/4b86661ee3db4afdd8df6614b645ec5a138997e9"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-05-02 14:30:20.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2024-08-24 17:16:44.000000000","tz":0},"subject":"Add pre-commit configuration","message":"Add pre-commit configuration\n\n\u0027pre-commit\u0027 tool [1] provides an extensible way to configure hooks\nwhich can be run automatically before each commit. That way we can make\nsure that code send to review is passing base static analysis checks\nlike e.g. bandit, flake8 or hacking.\n\nFor now this is optional to use and not integrated in e.g. tox.ini file\nbut we can integrate it there too as next step.\n\n[1] https://pre-commit.com/\n\nChange-Id: I9801bc7f23add5d4540913ce12eab1716dec62e5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/fb51aa5ccbd279f69dfa978e2a4dcdffece865e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/fb51aa5ccbd279f69dfa978e2a4dcdffece865e9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
