)]}'
{"id":"openstack%2Fnetworking-sfc~920868","triplet_id":"openstack%2Fnetworking-sfc~master~I17d2e0e391cd61cfa913004ac40a8aa26b10af18","project":"openstack/networking-sfc","branch":"master","topic":"fix_str_url-patch1","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-04 16:51:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I17d2e0e391cd61cfa913004ac40a8aa26b10af18","subject":"Switch back to attribute_mapped_collection","status":"ABANDONED","created":"2024-05-31 07:43:52.000000000","updated":"2024-08-04 16:51:33.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"809b4876c12d95f52813f5071bbfdd1ecb65aca4","_number":920868,"virtual_id_number":920868,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-31 16:31:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-31 11:00:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-04 16:51:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"}],"messages":[{"id":"08f8f47d78ef316067940f0d9529b3dd8222fa91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 07:43:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb185a6ddfb02241c223050cebcf2a4f34ab2f80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 11:00:58.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efa5956e88bd4247939f7885cdb25092\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c674ba3ba5494ca1e28f60f826ec8a : FAILURE in 9m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/943d27263984450a93bc58ce2a259e8b : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe9ee87118ed41bab3985b961a3f6b34 : SUCCESS in 6m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90094613fde64c938ef8ef7ba4565e02 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eaa5e50e69e412698c72dec61e39dc6 : SUCCESS in 8m 46s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3542fbae4bcd4318a130deafc61c1de8 : SUCCESS in 1m 45s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/f9774c8300d741069261e2a45ec32940 : SUCCESS in 14m 02s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/104abe4839a2456180b55884557d2749 : FAILURE in 2h 46m 51s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/a989f7b268804bde920f7dfd7da06694 : FAILURE in 2h 55m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55be7b0eae5566e683d6536cb224285e3eec2c3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 11:53:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7efafea9490b8498e1ea0b365b4c8bf6438bd1b5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 12:15:04.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e56fefba11ef2a335247f3a1744a44cdca67f2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 13:11:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"688b396622fdadf54414636f9a29f3908ad2b72a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 13:23:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f3e5f3caaa224beb6ea498d11eaa362b2e358e4d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 13:47:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"54f40a6537dc79ce57413c582cfe2a02688ba600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 16:31:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68a71b8cce844f7cabce72d2ebc6a8d5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5735552db2644318def0a843237671c : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e89a82e8e9743faa9c424fd8928fb15 : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae6cd250fa624f6896983894e7e256fa : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46ecd7a13c8349b085724a8854878ad3 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aae26a5895874946992b0ad18a07a10f : SUCCESS in 7m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7100130aa83c452cbaf04d568c731618 : SUCCESS in 1m 58s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/4b230ff5148d49e89ac5484c2e5bfa11 : SUCCESS in 22m 34s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/7151f67fb43b48f586471e68ca5ca02d : FAILURE in 3h 02m 19s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/c6bb90741e4e4c2bb4e97bbb6e7636e7 : FAILURE in 2h 55m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b441a89267129b950362839df0f669cdeb70860","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-31 17:18:12.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e2131d8e56c513ca910ce84e94a71fe470b281e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 20:31:32.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fa676b3c8e64e7da734b7e707362342\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a77896e8ab24811a6462e8ef5158bde : SUCCESS in 10m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b58bdf2f9c094a729b9bfd88f952e6e5 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ecd7386d06524d358a3bd89b985f9dc5 : SUCCESS in 6m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b746a0b725c54bc6be8894afa06e7a78 : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3049d91226647e9a76a2ae283b98978 : SUCCESS in 6m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cc15d5cbec364abe8c38608b57f480ae : SUCCESS in 2m 04s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/a2ead188266d4797a229d6fd30771c38 : SUCCESS in 22m 47s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/553dee410dfa4ee6883c40338372b445 : FAILURE in 3h 01m 18s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/23670b5a295340ce8f4a9eee50db9b90 : FAILURE in 2h 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"667f55b47e02faeab6bc4a1a4dab27528f483e6b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-03 09:09:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9fbe60511836a0eb95790a11553704a066ddafe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 12:14:53.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/da829419cdd04ed6b6967e9b7ddb5c1c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4416d8742e8841bfb08dcc2b557872d2 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a20fe251e5b4c3ea0be9b7c207fc0f4 : SUCCESS in 7m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/684ebc3acdfe4c609995849aee5d493b : SUCCESS in 8m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9bb18fa961c44506a669e83fa94a8857 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/306bce576f4e4995abe1497b425b6176 : SUCCESS in 8m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f7d947586b54e5ab5bb29ed132eaafa : SUCCESS in 2m 00s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/1f943122268343d2b787a3b9f68e3e27 : SUCCESS in 22m 40s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/b0bef9f42d8e4f83a9c98862f40f583c : FAILURE in 2h 59m 19s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/b3eac318150a4ed2add7d517399a4d92 : FAILURE in 2h 46m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"809b4876c12d95f52813f5071bbfdd1ecb65aca4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-08-04 16:51:33.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5ebb70e8a9890eec5eb7187b41fffed2db97de90","revisions":{"adeb824f158d43ab66c7261438a1dc43aec71aa7":{"kind":"REWORK","_number":1,"created":"2024-05-31 07:43:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/920868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/68/920868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/1"}}},"commit":{"parents":[{"commit":"bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18","subject":"Merge \"Add py312 job to periodic based on testing runtime for 2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-31 07:41:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-31 07:43:37.000000000","tz":540},"subject":"Switch back to attribute_mapped_collection","message":"Switch back to attribute_mapped_collection\n\nThis is partial revert of I8419f508f0185d2a896f5fc801145f945181d999 .\n\n2024.2 release has to support Ubuntu 22.04 according to tested runtimes\nand Ubuntu 22.04 needs SQLAlchemy \u003c 2.\n\nThe old method is still available in the latest SQLAlchemy version.\n\nChange-Id: I17d2e0e391cd61cfa913004ac40a8aa26b10af18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/adeb824f158d43ab66c7261438a1dc43aec71aa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/adeb824f158d43ab66c7261438a1dc43aec71aa7"}]},"branch":"refs/heads/master"},"692ff33bec900fc136cb2eb73c037ed26868915e":{"kind":"REWORK","_number":2,"created":"2024-05-31 11:53:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/920868/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/68/920868/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/2"}}},"commit":{"parents":[{"commit":"bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18","subject":"Merge \"Add py312 job to periodic based on testing runtime for 2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-31 07:41:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-31 11:53:23.000000000","tz":540},"subject":"Switch back to attribute_mapped_collection if needed","message":"Switch back to attribute_mapped_collection if needed\n\n2024.2 release has to support Ubuntu 22.04 according to tested runtimes\nand Ubuntu 22.04 needs SQLAlchemy \u003c 2.\n\nChange-Id: I17d2e0e391cd61cfa913004ac40a8aa26b10af18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/692ff33bec900fc136cb2eb73c037ed26868915e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/692ff33bec900fc136cb2eb73c037ed26868915e"}]},"branch":"refs/heads/master"},"ea109502ba4e881b354d5fc141801189022d8b3f":{"kind":"REWORK","_number":3,"created":"2024-05-31 13:11:09.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/920868/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/68/920868/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/3"}}},"commit":{"parents":[{"commit":"bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18","subject":"Merge \"Add py312 job to periodic based on testing runtime for 2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-31 07:41:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-31 13:10:26.000000000","tz":540},"subject":"Switch back to attribute_mapped_collection","message":"Switch back to attribute_mapped_collection\n\nThis is partial revert of I8419f508f0185d2a896f5fc801145f945181d999 .\n2024.2 release has to support Ubuntu 22.04 according to tested runtimes\nand Ubuntu 22.04 needs SQLAlchemy \u003c 2. The old method is still\navailable in the latest SQLAlchemy version and should be used now to\nkeep compatibility with SQLAlchemy \u003c 2.\n\nChange-Id: I17d2e0e391cd61cfa913004ac40a8aa26b10af18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ea109502ba4e881b354d5fc141801189022d8b3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/ea109502ba4e881b354d5fc141801189022d8b3f"}]},"branch":"refs/heads/master"},"5ebb70e8a9890eec5eb7187b41fffed2db97de90":{"kind":"REWORK","_number":4,"created":"2024-05-31 13:23:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/68/920868/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/68/920868/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/68/920868/4"}}},"commit":{"parents":[{"commit":"bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18","subject":"Merge \"Add py312 job to periodic based on testing runtime for 2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-31 07:41:21.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-31 13:23:38.000000000","tz":540},"subject":"Switch back to attribute_mapped_collection","message":"Switch back to attribute_mapped_collection\n\nThis is partial revert of I8419f508f0185d2a896f5fc801145f945181d999 .\n2024.2 release has to support Ubuntu 22.04 according to tested runtimes\nand Ubuntu 22.04 needs SQLAlchemy \u003c 2. The old method is still\navailable in the latest SQLAlchemy version and should be used now to\nkeep compatibility with SQLAlchemy \u003c 2.\n\nChange-Id: I17d2e0e391cd61cfa913004ac40a8aa26b10af18\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5ebb70e8a9890eec5eb7187b41fffed2db97de90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5ebb70e8a9890eec5eb7187b41fffed2db97de90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
