)]}'
{"id":"openstack%2Fnetworking-sfc~921514","triplet_id":"openstack%2Fnetworking-sfc~master~I980e89d7cee5b2fd0c74670e16805096c9d0ec36","project":"openstack/networking-sfc","branch":"master","topic":"bug/2068727","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-08-23 14:01:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I980e89d7cee5b2fd0c74670e16805096c9d0ec36","subject":"Handle missing vlan for egress flow setup","status":"ABANDONED","created":"2024-06-07 12:26:52.000000000","updated":"2024-08-23 14:01:14.000000000","total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"d8f2d31c62da1165bdbdf8952b7a3b59011814fb","_number":921514,"virtual_id_number":921514,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-11 14:45:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-07 15:23:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-12 15:51:03.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"CC"}],"messages":[{"id":"07ab0a754e941e67858106b11f07d61dea58ac71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-06-07 12:26:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66b788fae9f9aca573469b6b0ad37e97113fa6bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 15:23:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be208be9be5544babb9c96c83ace37de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8ccbb21562e496eb8458516cd3e41e7 : SUCCESS in 10m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c8236322e7547a2ad0698d125fde477 : FAILURE in 6m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4003e46b97084441bc8d29ce88ef52e6 : FAILURE in 7m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3111e23f0d824ab9afc151004cdff4be : FAILURE in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58982aab35ba44a1b6c1c6eda3f6ed24 : SUCCESS in 6m 52s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/3b63748fb84f4f3ca3b4399924e98f5d : SUCCESS in 12m 06s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/73623f1916954f2e95fb06d66f26fccd : FAILURE in 2h 43m 17s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/43fbb59ef6314a68956d6f014b8e2c37 : FAILURE in 2h 47m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"348d5cbee87324a6640096c2a597c5f383738807","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-06-11 11:37:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"70b799260408999b51e480fc1f2d55928c587ae3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 14:45:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/019fce6208e34c82a71ab0a18c8809af\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/611e937d1fdb434d91bff6883ec56c74 : SUCCESS in 10m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb97fae1beb0460ebe339b7f3aafcc97 : SUCCESS in 6m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44a4f965c88a408f845bb0c521f3d040 : SUCCESS in 6m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/313f8750ffc740948ac4b9dc7ac9953a : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a278ea6379234f69acdba04cc081e4b5 : SUCCESS in 8m 41s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/3939ef524be3455ab670ce21b8167ec5 : SUCCESS in 22m 51s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/812e62c7dac242e39f79a122349f40cd : FAILURE in 2h 52m 57s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/8bdac8c4bbab41fba6bba1a399fef8b7 : TIMED_OUT in 3h 02m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24b39c1ae17932bb45e22cbf0e75ee6a04208dc8","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2024-06-12 15:51:03.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"90e65d2e5d04a4c06ce1cb76c295ae49c789284f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-18 10:37:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b6baa017feb3c6f2bba89f8972bce0a66e35182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 13:31:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2546942ad549436fbba7f8da45124bd5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/133a673185bd43408423e89bd2f4c849 : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3aeb9bdff4cd44e788f8c3a59ace3bf5 : SUCCESS in 7m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80c60d13590f450c9f60404a7e08881a : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf0bf7f565464c5f8637a76d2ea9ba65 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343a2b4db72e418c817e1fff420541e0 : SUCCESS in 9m 30s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/a0dbc70603794e6488ec2c93e286adfa : SUCCESS in 10m 52s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/5159fda5786046a3af7c02560ad84199 : FAILURE in 2h 53m 30s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/e1ced8e01b1e44b6896e9c4e59466e55 : FAILURE in 2h 44m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d8f2d31c62da1165bdbdf8952b7a3b59011814fb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-23 14:01:14.000000000","message":"Abandoned\n\nSlaweq has a working solution: https://review.opendev.org/c/openstack/networking-sfc/+/926892?usp\u003dsearch","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"92fe2f7a50b01bb080c17687c438490d472fbce7","revisions":{"bda4a686c4a5c4d57a132f8dc1c3b80a7aa4ce51":{"kind":"REWORK","_number":1,"created":"2024-06-07 12:26:52.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/14/921514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/14/921514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/921514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/921514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/921514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/14/921514/1"}}},"commit":{"parents":[{"commit":"bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18","subject":"Merge \"Add py312 job to periodic based on testing runtime for 2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-06-07 12:24:13.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-06-07 12:24:13.000000000","tz":120},"subject":"Handle missing vlan for egress flow setup","message":"Handle missing vlan for egress flow setup\n\nIn _setup_egress_flow_rules (in sfc_driver.py)\nvlan can be mepty returned from vlan manager. Handle\nthis case and log some useful information.\n\nRelated-Bug: #2068727\nChange-Id: I980e89d7cee5b2fd0c74670e16805096c9d0ec36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bda4a686c4a5c4d57a132f8dc1c3b80a7aa4ce51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bda4a686c4a5c4d57a132f8dc1c3b80a7aa4ce51"}]},"branch":"refs/heads/master"},"92fe2f7a50b01bb080c17687c438490d472fbce7":{"kind":"REWORK","_number":2,"created":"2024-06-11 11:37:06.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/14/921514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/14/921514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/921514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/921514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/14/921514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/14/921514/2"}}},"commit":{"parents":[{"commit":"bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18","subject":"Merge \"Add py312 job to periodic based on testing runtime for 2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/bcc6be3e8b4e89ad6bb6bff66d2c537ee8c77f18"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-06-07 12:24:13.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-06-11 11:36:53.000000000","tz":120},"subject":"Handle missing vlan for egress flow setup","message":"Handle missing vlan for egress flow setup\n\nIn _setup_egress_flow_rules (in sfc_driver.py)\nvlan can be mepty returned from vlan manager. Handle\nthis case and log some useful information.\n\nRelated-Bug: #2068727\nChange-Id: I980e89d7cee5b2fd0c74670e16805096c9d0ec36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/92fe2f7a50b01bb080c17687c438490d472fbce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/92fe2f7a50b01bb080c17687c438490d472fbce7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
