)]}'
{"id":"openstack%2Fnetworking-sfc~929884","triplet_id":"openstack%2Fnetworking-sfc~master~I94ec7343b809347543c836543059b4602beeb46f","project":"openstack/networking-sfc","branch":"master","topic":"remove_sqlalchemy_20_main_jobs","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2024-09-19 14:12:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I94ec7343b809347543c836543059b4602beeb46f","subject":"[sqlalchemy-20] Remove CI job \"openstack-tox-py310-with-sqlalchemy-main-sfc\"","status":"MERGED","created":"2024-09-19 08:03:00.000000000","updated":"2024-09-19 14:15:41.000000000","submitted":"2024-09-19 14:12:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929884-remove_sqlalchemy_20_main_jobs","meta_rev_id":"fab287ed6fde405fde559d9d8c024556cf02f911","_number":929884,"virtual_id_number":929884,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-19 14:12:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":2,"date":"2024-09-19 12:17:02.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":2,"date":"2024-09-19 08:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"all":[{"value":1,"date":"2024-09-19 12:17:02.000000000","_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-19 08:46:25.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-09-19 10:14:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-19 12:17:02.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"}],"messages":[{"id":"c1ac8b8a80cad0ffca4965d47c77764df7832a71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-19 08:03:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d11fe5e42776dc4154d37c21006485b803b65038","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-19 08:46:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0095f6c0a145da36c5f4c8624bb560fc42b1c6cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 10:14:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3310913ed094a518e5b1c49d2e4b5d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93dc7762a164fb489710f5981611792 : SUCCESS in 10m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b764650b0284244867e3b4148879814 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d15ccff3fedf44b48d8851ab26009a27 : SUCCESS in 5m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/37954d725f174262851846a142c3e5b7 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6a6c574fd834f068a168a1570cc41d4 : SUCCESS in 8m 14s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/021561747ca843dda7e4b0b631bab56e : SUCCESS in 24m 25s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/4f8acf59ce6c4203ab5d5bf77e7b284a : SUCCESS in 1h 39m 20s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/12cc1aee954e486f80cbd0b9f577071a : SUCCESS in 2h 05m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88363fefc3590dde9f614743dd75c8c404f6c0a7","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2024-09-19 12:17:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"578e2d99e854f2fd336fec45cdd8ba6f675e1fbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 12:17:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c342f6df215602f2024c9fee866b8b481720dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 14:12:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8451e3438cf46bd8bd437279a913e75\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5404afe72b914ea288e369e5f3bfe804 : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ce0aa2be974446494e76ef2a7d49b21 : SUCCESS in 7m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/badc438f272a4d4d93b0d0ade7987a51 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e60906e656f49f2b564abbc5865f341 : SUCCESS in 7m 55s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/d9c207b1b2ba4b9a997c2e902ea6f66d : SUCCESS in 17m 08s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/d872e75706d84d5296671b6e1a939a1e : SUCCESS in 1h 50m 10s","accounts_in_message":[],"_revision_number":1},{"id":"ea04673029065d31012e3bf0dc59bcaf50610642","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 14:12:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fab287ed6fde405fde559d9d8c024556cf02f911","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 14:15:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e47f590779994a9484549f4af36fd79a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dddbda9444a4347960ccf495ec69d26 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15f6ba9a74e75437a8cf61a845d59fc3cc18c138","revisions":{"15f6ba9a74e75437a8cf61a845d59fc3cc18c138":{"kind":"REWORK","_number":1,"created":"2024-09-19 08:03:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/84/929884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/84/929884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/84/929884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/84/929884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/84/929884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/84/929884/1"}}},"commit":{"parents":[{"commit":"3c4e6473f8ee42e9ff1df3630f55a43e7dc5d4da","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/3c4e6473f8ee42e9ff1df3630f55a43e7dc5d4da"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-19 08:00:26.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2024-09-19 08:02:27.000000000","tz":0},"subject":"[sqlalchemy-20] Remove CI job \"openstack-tox-py310-with-sqlalchemy-main-sfc\"","message":"[sqlalchemy-20] Remove CI job \"openstack-tox-py310-with-sqlalchemy-main-sfc\"\n\nThe SQLAlchemy 2.0 migration is concluded and there is no need to keep\nthe jobs testing the main branch anymore.\n\nChange-Id: I94ec7343b809347543c836543059b4602beeb46f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/15f6ba9a74e75437a8cf61a845d59fc3cc18c138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/15f6ba9a74e75437a8cf61a845d59fc3cc18c138"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
