)]}'
{"id":"openstack%2Fnetworking-sfc~935611","triplet_id":"openstack%2Fnetworking-sfc~master~I5fdad31a62c3a64fd42fa74b359691283dc7bf53","project":"openstack/networking-sfc","branch":"master","topic":"test_db_base_plugin_v2","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-11-19 13:02:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5fdad31a62c3a64fd42fa74b359691283dc7bf53","subject":"tests: test_db_base_plugin_v2.py was moved to common","status":"MERGED","created":"2024-11-19 09:01:22.000000000","updated":"2024-11-19 13:03:51.000000000","submitted":"2024-11-19 13:02:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935611-test_db_base_plugin_v2","meta_rev_id":"5bdb6ed092e54e0f4ee29e06cb9faf25a0af7496","_number":935611,"virtual_id_number":935611,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-19 13:02:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2024-11-19 09:15:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2024-11-19 09:15:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 09:15:14.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2024-11-19 11:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c15b0a6103ab6e018962a09543f5043cc04d3c67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-11-19 09:01:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5318bb648afd514bba4a89f8794b6fc93ca7da81","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-11-19 09:15:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6043bdc46607798940ed72f9f6b697e41a37558d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 11:07:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd18d9e1eb064617b17535606f611c01\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/319a277165614f5d9486ba0eaf182bb3 : SUCCESS in 10m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c2ddebe515b474fa65e6b30b3208603 : SUCCESS in 8m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f313a0b182c341248e7d3322aba1e26c : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd42db9094d4a3aa93d21600c87ad2d : SUCCESS in 7m 32s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/63647858777e4d03b0610467bc54d7c3 : SUCCESS in 23m 22s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/68f6ff87947a4010af6281669202ac6b : SUCCESS in 1h 50m 38s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/e4252577f2b64a90b4cce24e62f003c5 : SUCCESS in 2h 00m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c0ab8bfdd4dc907a0a155276863c192a11ffb04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 11:10:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dcf60a05ec25890e161dd647d19b5c79e35374e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:02:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e911fdb971c45cd8c44ef558beacdb6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41f673cca26c4f3497914b02437205a0 : SUCCESS in 10m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1889872b919c4e44a49286855da59889 : SUCCESS in 6m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7515935720bd4f5da5816ac2902ef6e6 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc8f841cf9a45e0b048737a661a8d54 : SUCCESS in 7m 31s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/19009ea748b64eedbfbdd70f223d6b84 : SUCCESS in 24m 02s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/e004e278e699416c8175736ae3b1c6d1 : SUCCESS in 1h 51m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bd6d1bd849038f63c59c01a7461f0c6517f753a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:02:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5bdb6ed092e54e0f4ee29e06cb9faf25a0af7496","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 13:03:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73c4292b3fba424b90470dd6629a4442\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09937e58996a46abac54d552c16bb671 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5fd80d4cf01e55232f43883a61a74bd3c98fd0d","revisions":{"f5fd80d4cf01e55232f43883a61a74bd3c98fd0d":{"kind":"REWORK","_number":1,"created":"2024-11-19 09:01:22.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/11/935611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/11/935611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/935611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/935611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/11/935611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/11/935611/1"}}},"commit":{"parents":[{"commit":"43b8889dc9f2395c48259fb5e845b1e293e1d5cd","subject":"reno: Update master for unmaintained/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/43b8889dc9f2395c48259fb5e845b1e293e1d5cd"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-19 08:56:16.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-11-19 08:56:16.000000000","tz":60},"subject":"tests: test_db_base_plugin_v2.py was moved to common","message":"tests: test_db_base_plugin_v2.py was moved to common\n\nIn [1] test_db_base_plugin_v2.py was moved to tests/common\nlet\u0027s adapt to it.\n\n[1]: https://review.opendev.org/c/openstack/neutron/+/934190\n\nChange-Id: I5fdad31a62c3a64fd42fa74b359691283dc7bf53\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f5fd80d4cf01e55232f43883a61a74bd3c98fd0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f5fd80d4cf01e55232f43883a61a74bd3c98fd0d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
