)]}'
{"id":"openstack%2Fnetworking-sfc~936253","triplet_id":"openstack%2Fnetworking-sfc~master~I509928d3c1c531af48a809b66d5b37d33be214b3","project":"openstack/networking-sfc","branch":"master","topic":"bug/2089673","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-12-02 11:49:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I509928d3c1c531af48a809b66d5b37d33be214b3","subject":"[DNM] check sfc functional with the fix","status":"ABANDONED","created":"2024-11-26 12:41:25.000000000","updated":"2024-12-02 11:49:51.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c7a400fda800228226d268d56d73d16218e83c37","_number":936253,"virtual_id_number":936253,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-02 11:47:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 13:21:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4af648d524932cf442735e0e94841b3d9dd37055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-11-26 12:41:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c29a0e74abc9bb089542a08f951a14e2ba8e0950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 13:21:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c883df0351694b178580bd6865a9f9b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42fd439c5d454b069ed07c2d3a06278f : SUCCESS in 10m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a2c8414e8874dba9aac827d57d901a4 : SUCCESS in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f05afda552d348eebbefcab82c1b111a : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b61f0f98877e4f909a6f1676db65d9b6 : SUCCESS in 6m 44s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/1747f76e31e74081902f90dacf058d18 : FAILURE in 39m 33s","accounts_in_message":[],"_revision_number":1},{"id":"12d119ff3edcdc546fb7d14635a90bd0ef04af19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-11-26 13:39:51.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"20666c953903318032d078779f182e76647c2872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 14:12:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a0f96b09b7c46f79040809ea31a386e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd906ad364b483a8636ca5daf605f62 : SUCCESS in 8m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0899ee623a554d1091577d42997015e0 : SUCCESS in 7m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2eda6ba3d3d64ac1a4d6bcdaa6c1e612 : SUCCESS in 9m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2899678384104a31afdc0909631aad80 : SUCCESS in 8m 45s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/ac7795f77e87408d891aeaccda0c0040 : SUCCESS in 25m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b336f75267bf7764b866861c443a54cc1a53e4ed","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-11-27 06:05:13.000000000","message":"Abandoned\n\nfix merged","accounts_in_message":[],"_revision_number":2},{"id":"66abf276dde01d8c993783ae7da13a3ef04e55a7","tag":"autogenerated:gerrit:restore","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-02 11:02:20.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"1dd4499526405e73b165c0ea46cfdaec54f30477","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-02 11:02:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0f45753dc9e6238bb4ee6b6e46521f920175028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 11:47:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71ed79f2ab934855aad218ee225c9840\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b7304b911324e3aba0904ec589312ab : FAILURE in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f0f83c3b4384dd989d589cb699be3b6 : SUCCESS in 6m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e5b92901f62483a957f0867d84c795f : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12331383144c47819b840cfb536abb96 : SUCCESS in 8m 35s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/c955975cfe354ad9942a4ec836de4a28 : SUCCESS in 28m 57s\n\nWarning:\n  Comments left for invalid file networking_sfc/db/flowclassifier_db.py\n  Comments left for invalid file networking_sfc/db/sfc_db.py\n  Comments left for invalid file networking_sfc/services/sfc/drivers/ovs/db.py\n  Comments left for invalid file networking_sfc/services/sfc/drivers/ovs/driver.py\n  Comments left for invalid file networking_sfc/services/sfc/drivers/ovs/rpc.py","accounts_in_message":[],"_revision_number":2},{"id":"c7a400fda800228226d268d56d73d16218e83c37","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-12-02 11:49:51.000000000","message":"Abandoned\n\nfunctional green now with neutron noble switch patch merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8c85dbea4b747575c56f48dbc51172c371495e2","revisions":{"c2c0892f133b304f8c880f2e0b2edc196d9484c2":{"kind":"REWORK","_number":1,"created":"2024-11-26 12:41:25.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/53/936253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/53/936253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/53/936253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/53/936253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/53/936253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/53/936253/1"}}},"commit":{"parents":[{"commit":"f5fd80d4cf01e55232f43883a61a74bd3c98fd0d","subject":"tests: test_db_base_plugin_v2.py was moved to common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f5fd80d4cf01e55232f43883a61a74bd3c98fd0d"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-26 12:39:29.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-26 12:41:21.000000000","tz":330},"subject":"[DNM] check sfc functional","message":"[DNM] check sfc functional\n\nSeen failure in weekly run, checking consistency.\n\nChange-Id: I509928d3c1c531af48a809b66d5b37d33be214b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c2c0892f133b304f8c880f2e0b2edc196d9484c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/c2c0892f133b304f8c880f2e0b2edc196d9484c2"}]},"branch":"refs/heads/master"},"e8c85dbea4b747575c56f48dbc51172c371495e2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-26 13:39:51.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/53/936253/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/53/936253/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/53/936253/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/53/936253/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/53/936253/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/53/936253/2"}}},"commit":{"parents":[{"commit":"f5fd80d4cf01e55232f43883a61a74bd3c98fd0d","subject":"tests: test_db_base_plugin_v2.py was moved to common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f5fd80d4cf01e55232f43883a61a74bd3c98fd0d"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-26 12:39:29.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2024-11-26 13:38:54.000000000","tz":330},"subject":"[DNM] check sfc functional with the fix","message":"[DNM] check sfc functional with the fix\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/936262\nChange-Id: I509928d3c1c531af48a809b66d5b37d33be214b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e8c85dbea4b747575c56f48dbc51172c371495e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e8c85dbea4b747575c56f48dbc51172c371495e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
