)]}'
{"id":"openstack%2Fnetworking-sfc~937018","triplet_id":"openstack%2Fnetworking-sfc~master~Ia6ae2a2891d6cedfe6fc497623978959e03080ba","project":"openstack/networking-sfc","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-12-06 12:29:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia6ae2a2891d6cedfe6fc497623978959e03080ba","subject":"DNM: test master","status":"ABANDONED","created":"2024-12-04 13:29:45.000000000","updated":"2024-12-06 12:29:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a79de1e5074c3e0b65c6b3c46d4f896c74c38f56","_number":937018,"virtual_id_number":937018,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-12-04 15:19:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-04 15:19:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"28a3c84fac7ecaddcd6f36242da3855f2dfdcfb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-04 13:29:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06038f6d00da0a08ab9dc930301598e1a0aa008a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-04 15:19:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92bc7945834242c7974805a1d12990b8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e36913cd78348818943701f48b22274 : FAILURE in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/165b0ab0e63e4ca19864f870c263ff71 : SUCCESS in 6m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/59bdedf9d6d74e4fb030f767e34ee6a4 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ac30da5da549b1bcd4bfd04d3ad3da : SUCCESS in 9m 20s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/977b247cd5f1460cba64b08a568f36f7 : SUCCESS in 18m 57s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/f39e9a5a71ce418da091409ca6dc6e8c : SUCCESS in 1h 42m 25s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/9910510c64c84c839ba5b17c551ceb4c : SUCCESS in 1h 41m 33s (non-voting)\n\nWarning:\n  Comments left for invalid file networking_sfc/db/flowclassifier_db.py\n  Comments left for invalid file networking_sfc/db/sfc_db.py\n  Comments left for invalid file networking_sfc/services/flowclassifier/plugin.py\n  Comments left for invalid file networking_sfc/services/sfc/drivers/ovs/db.py\n  Comments left for invalid file networking_sfc/services/sfc/drivers/ovs/driver.py\n  Comments left for invalid file networking_sfc/services/sfc/drivers/ovs/rpc.py\n  Comments left for invalid file networking_sfc/services/sfc/plugin.py","accounts_in_message":[],"_revision_number":1},{"id":"a79de1e5074c3e0b65c6b3c46d4f896c74c38f56","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-12-06 12:29:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a20addb171ce0e86a5091a1986425d9ef038d2a","revisions":{"5a20addb171ce0e86a5091a1986425d9ef038d2a":{"kind":"REWORK","_number":1,"created":"2024-12-04 13:29:45.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/18/937018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/18/937018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/18/937018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/18/937018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/18/937018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/18/937018/1"}}},"commit":{"parents":[{"commit":"f5fd80d4cf01e55232f43883a61a74bd3c98fd0d","subject":"tests: test_db_base_plugin_v2.py was moved to common","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/f5fd80d4cf01e55232f43883a61a74bd3c98fd0d"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-12-04 13:29:01.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-12-04 13:29:01.000000000","tz":60},"subject":"DNM: test master","message":"DNM: test master\n\nChange-Id: Ia6ae2a2891d6cedfe6fc497623978959e03080ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5a20addb171ce0e86a5091a1986425d9ef038d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/5a20addb171ce0e86a5091a1986425d9ef038d2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
