)]}'
{"id":"openstack%2Fnetworking-sfc~967783","triplet_id":"openstack%2Fnetworking-sfc~master~Ib93a2bf2e32ab4e5779c1d91c08b5ebc73ea69c8","project":"openstack/networking-sfc","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-12 08:58:59.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-01-12 12:55:33.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2026-01-12 12:55:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib93a2bf2e32ab4e5779c1d91c08b5ebc73ea69c8","subject":"Change the vxlan vni_range to be 1:1000","status":"MERGED","created":"2025-11-19 22:17:53.000000000","updated":"2026-01-12 12:57:11.000000000","submitted":"2026-01-12 12:55:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"967783","meta_rev_id":"d18a7c22c5613bc6d445cb018eff3e8f580608d4","_number":967783,"virtual_id_number":967783,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-12 12:55:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-01-12 08:58:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-01-12 08:58:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2026-01-12 08:58:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-19 23:20:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-21 20:56:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2025-11-25 22:47:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-11-27 23:22:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2026-01-07 21:35:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"f875ec067a2b2a4ecedf4ec47d071dcc64691af0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-19 22:17:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"341f278472a54264d2e6a08379c2c2a437e25f3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-19 23:20:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be8023a9050f4cd5a73680ef6a003061\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a8e162b1b44ae98b696ca15e4fe60d : SUCCESS in 7m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebdc6d01a8844520b18b46a1d1a33b4a : SUCCESS in 7m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d4f6e33d467444899f08b740f03d783 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33c0c42026b4aa984439f1139dc591e : SUCCESS in 10m 20s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/915167a4ae754e60aef75d189809a173 : SUCCESS in 23m 30s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/370aaba111cc43b89874caf960c1094b : FAILURE in 1h 00m 13s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/8d43d85bc9aa43738cea8a2699acffa7 : FAILURE in 58m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27e330f5582a0994ed012db295224a018406b227","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-19 23:21:33.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae2eef70e5d0f6c2cdf663221f840f77901c252d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2025-11-19 23:52:47.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"49fb2539e676df04a9c308df82163274d58ee106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-20 01:39:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d73bb56cbb004fc6a1bcb616f2fdeb32\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b541d856050e4ccbb138ef841c986006 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef183d30fb5d47dab71b10231fbc906c : SUCCESS in 6m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb334cbf6b5d4e469114b2e3401b0adf : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6537f272548d482ebd16bb4ddc0e5817 : SUCCESS in 11m 03s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/249fdb73af004bc38786ddc08712b2d6 : SUCCESS in 13m 18s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/3e49300fb4064c52a972d3192a0f04c0 : FAILURE in 1h 22m 01s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/1735b1b6f6c2443db3344f2ad01abfdf : FAILURE in 1h 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a187c05e8f43e9118ecef34365d5f70b13e7207e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-21 20:56:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9b99abb33776eb8da2b68b7b4c090d7afcbaa437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 22:04:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cfa4009ef9f49c6a9e362053f508865\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc8d457b8cd43adbc087f7b561a0f49 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d49c467b37d4623b1564aaaa2427550 : SUCCESS in 6m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1faea4fc704c4ab38ae86df36a13f573 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c272ac9987854dee86beec82dd0c3cf3 : SUCCESS in 6m 11s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/f30c41e116564785a2ea5e2462f6d32d : SUCCESS in 14m 51s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/7a8762c9742e459db5234089005a4d55 : SUCCESS in 1h 05m 41s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/22ad445752a24b2aa4d1b6f47f39db4a : SUCCESS in 1h 03m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65a23162deb61e36dc47133770b71b8263b01047","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-11-25 22:47:12.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0556cea0050ccdfd36dac326c12d93939fa68aa8","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2025-11-27 23:22:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"17444c89aad7be46f1a2d327d08e655a7c932ed3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-08 08:17:38.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2b8add015e3e17d5de57ad3b5a62885ed28bfb8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 08:18:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"53fd3e61d537a44457d627da9a8ae27a591fb2e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 10:09:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f207d821a7bb4eeea8e322eabb0a8b47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ada5085e31e4ef68fc80850e77cc6a7 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93c878cae16a44d296070d10adbaf635 : FAILURE in 6m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cff739e7dd31496188c87aeaf3c203b6 : FAILURE in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6470e0212f19441294cb39464b380672 : SUCCESS in 4m 37s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/931dadb4efb048c49ca553f6debae194 : SUCCESS in 9m 50s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/5b2f88c524fc48f0821beaf89ba9c1c8 : SUCCESS in 1h 50m 19s","accounts_in_message":[],"_revision_number":2},{"id":"3cf06e5a070d045c5f90da7bef34acd71d895e96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"real_author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-12 08:58:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_1131\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}],"_revision_number":3},{"id":"44bbdb2d67a14cb9f4c49d3e9291517237f01cdd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-01-12 08:58:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cbdab031b6197e92981ed8f206b8d456c8036e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 11:04:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c01c68e0a494f2db4b634bf4027ead3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/491167a873af4a6ab28583e5cc2d5535 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/861c5bef599d412483efe6cb161ec6fa : SUCCESS in 7m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dbf5d7649efa4493ba4a3a00147b59a0 : SUCCESS in 10m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac367d1aaf4644b29987eb9ee5cf5002 : SUCCESS in 10m 45s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/fc74d52d38224d23a785211c59c2e537 : SUCCESS in 29m 08s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/edd2fee645b0403987d5e3d9cdee7247 : SUCCESS in 1h 51m 10s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/e9f348ea81054b548eeade7a28e80d68 : SUCCESS in 1h 59m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a34c2e910463ac21cadfeeee53663cce736bed3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 11:05:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"127348607dbaee197f49f5e42810eca3e5f8da0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 12:55:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b3a5613282f4b9194bf4480411dcabe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f1fb6873444d15af92842cfde9aa96 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d3d90e0d665b45429eba65b2515dc061 : SUCCESS in 6m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f14787b0c81545df9ff6456aa1f2fefc : SUCCESS in 11m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be8be79a32d421d8f726eb48a36f1e9 : SUCCESS in 11m 39s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/9447203f846f4e118c978d45be082006 : SUCCESS in 11m 46s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/18a1f9e9eb5c4f288692100e536758da : SUCCESS in 1h 46m 47s","accounts_in_message":[],"_revision_number":3},{"id":"3227ab932d4ee370feb460af31cf9c85f408d873","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 12:55:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d18a7c22c5613bc6d445cb018eff3e8f580608d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 12:57:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e8c7651a7f343ef9e4cf0284e488cc4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a019c9bbfd4fda80bc568b23cd862a : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"58961470685e7d78f3e07163d7b69792928bd600","revisions":{"fb0d776a4e954808b71f02f6efad310fdcce5c03":{"kind":"REWORK","_number":1,"created":"2025-11-19 22:17:53.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/83/967783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/83/967783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/1"}}},"commit":{"parents":[{"commit":"d615b5d90cfd928999e633749a6848312e1e2800","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d615b5d90cfd928999e633749a6848312e1e2800"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-11-19 22:14:42.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-11-19 22:14:42.000000000","tz":-300},"subject":"Change the vxlan vni_range to be 1:1000","message":"Change the vxlan vni_range to be 1:1000\n\nThe multinode job definition uses 1:2000 for the vxlan\nvni range, change to 1:1000 to match what is done\nin neutron-tempest-plugin repo.\n\nChange-Id: Ib93a2bf2e32ab4e5779c1d91c08b5ebc73ea69c8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/fb0d776a4e954808b71f02f6efad310fdcce5c03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/fb0d776a4e954808b71f02f6efad310fdcce5c03"}]},"branch":"refs/heads/master"},"9b0e5b7990c054ccaf77cde1fc7c2e8461affa8e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-11-19 23:52:47.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/83/967783/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/83/967783/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/2"}}},"commit":{"parents":[{"commit":"d615b5d90cfd928999e633749a6848312e1e2800","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/d615b5d90cfd928999e633749a6848312e1e2800"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-11-19 22:14:42.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-11-19 23:52:16.000000000","tz":-300},"subject":"Change the vxlan vni_range to be 1:1000","message":"Change the vxlan vni_range to be 1:1000\n\nThe multinode job definition uses 1:2000 for the vxlan\nvni range, change to 1:1000 to match what is done\nin neutron-tempest-plugin repo.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/967782\n\nChange-Id: Ib93a2bf2e32ab4e5779c1d91c08b5ebc73ea69c8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/9b0e5b7990c054ccaf77cde1fc7c2e8461affa8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/9b0e5b7990c054ccaf77cde1fc7c2e8461affa8e"}]},"branch":"refs/heads/master"},"58961470685e7d78f3e07163d7b69792928bd600":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-12 08:58:56.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"real_uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/83/967783/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/83/967783/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/83/967783/3"}}},"commit":{"parents":[{"commit":"e55ee5ba0c505aa6424507f02914a7c60ff4efe9","subject":"Replace \"tenant_id\" with \"project_id\" in testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/e55ee5ba0c505aa6424507f02914a7c60ff4efe9"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2025-11-19 22:14:42.000000000","tz":-300},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2026-01-12 08:58:56.000000000","tz":0},"subject":"Change the vxlan vni_range to be 1:1000","message":"Change the vxlan vni_range to be 1:1000\n\nThe multinode job definition uses 1:2000 for the vxlan\nvni range, change to 1:1000 to match what is done\nin neutron-tempest-plugin repo.\n\nDepends-on: https://review.opendev.org/c/openstack/neutron/+/967782\n\nChange-Id: Ib93a2bf2e32ab4e5779c1d91c08b5ebc73ea69c8\nSigned-off-by: Brian Haley \u003chaleyb.dev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/58961470685e7d78f3e07163d7b69792928bd600"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/58961470685e7d78f3e07163d7b69792928bd600"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
