)]}'
{"id":"openstack%2Fnetworking-sfc~987371","triplet_id":"openstack%2Fnetworking-sfc~master~I38ba09c2d5e79649e8693e626cceac8268586e46","project":"openstack/networking-sfc","branch":"master","topic":"bug/2150864","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-05-06 15:19:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I38ba09c2d5e79649e8693e626cceac8268586e46","subject":"Enable name as filter for SFC API","status":"ABANDONED","created":"2026-05-05 14:40:02.000000000","updated":"2026-05-06 15:22:12.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3dc737b3b3424dda7f7cfe20b3b2f32b94410905","_number":987371,"virtual_id_number":987371,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-05-05 16:49:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 16:49:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-06 11:12:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"9212cee5e06809812ec642faa8a5ac21344b2d4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-05 14:40:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44b2eafeedbfd71494ca3e54a18cdfc569d7d2b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:49:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1696afa042674d8babecbd86e0388fae\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a98d24b99b8437baccf2cbd4800d9ff : SUCCESS in 7m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04efa5ceee17427b910d160cb2a5123a : FAILURE in 7m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97d2c3daade74f1582cb0077b690165e : FAILURE in 7m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/50339ba842fe46bf813f0174840ef25e : FAILURE in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/481186f46b2a4153815ea599b97e1b13 : SUCCESS in 9m 03s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/27f644402d2944c098c28f0126b8c502 : SUCCESS in 25m 34s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/11f9756ea2904cae87fe9f0bdb741c30 : SUCCESS in 1h 52m 35s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/731fb56de1a942d3be108ab261cb5f2f : FAILURE in 2h 04m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e72a05970f79ecf2c56c3b8123247409f3e97a53","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-06 11:54:42.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/networking-sfc/+/987469 to disable the filtering in sfc is quicker, after the gate is green we can move to fix the API","accounts_in_message":[],"_revision_number":1},{"id":"db252d114e39ccdf8afe692e09cec812fe3ee05d","tag":"autogenerated:gerrit:restore","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-06 12:55:43.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"a42b7ab0299464cb6114e6ec6d28eca65bb89612","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-05-06 12:56:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"090ea6131dac84bb19d5c39e4dd175614d4aeb7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 14:35:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8eca6effe5af42ce8c38b7bfd795ea47\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd0225286d9404096cac1f136a6269d : SUCCESS in 3m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a20ab6f724a244eebc178d8e29dd6eb4 : FAILURE in 5m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/580a77c3b6884bc5b3b014763be9c135 : FAILURE in 5m 40s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/de84581a81a34d5798257364588cb8e3 : FAILURE in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b1fa3943474d01ade48aed72e36157 : SUCCESS in 10m 43s\n- networking-sfc-functional https://zuul.opendev.org/t/openstack/build/233a8ac7d52b4183b977a1e8773ec2ec : SUCCESS in 24m 38s\n- networking-sfc-tempest https://zuul.opendev.org/t/openstack/build/99fdba569b214b4c98ee31fcd4fa7613 : SUCCESS in 1h 30m 02s\n- networking-sfc-tempest-multinode https://zuul.opendev.org/t/openstack/build/e8009d57c4b9451aa5e278f2e4954dea : FAILURE in 1h 38m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7edd801239d3ac2a5b6a5253758f23d8b2af82cc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-06 15:19:53.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/networking-sfc/+/987469 fixed the issue short term, we can enable filtering in n-lib for these APIs and make sfc use the API definitions from n-lib","accounts_in_message":[],"_revision_number":1},{"id":"3dc737b3b3424dda7f7cfe20b3b2f32b94410905","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-05-06 15:22:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"13da26ebc22953501ca0a7ff2c04e3bdd82b5e60","revisions":{"13da26ebc22953501ca0a7ff2c04e3bdd82b5e60":{"kind":"REWORK","_number":1,"created":"2026-05-05 14:40:02.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/71/987371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/networking-sfc","ref":"refs/changes/71/987371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/71/987371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/71/987371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/networking-sfc refs/changes/71/987371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/networking-sfc refs/changes/71/987371/1"}}},"commit":{"parents":[{"commit":"41ba6e9e2385a996a037e25760042f162ee39dac","subject":"Update master for stable/2026.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/41ba6e9e2385a996a037e25760042f162ee39dac"}]}],"author":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-04 14:55:08.000000000","tz":120},"committer":{"name":"lajoskatona","email":"lajos.katona@est.tech","date":"2026-05-05 14:39:43.000000000","tz":120},"subject":"Enable name as filter for SFC API","message":"Enable name as filter for SFC API\n\nThe patches pushed for #2142034 (see [1]) made filter\nvalidation enabled for all extensions. The networking-sfc\nextensions had no is_filter attribute, this patch\nadds that to the name field as that is tested in\nunit tests, so I assume was expected to work at least.\n\nNote: The networking-sfc API definitions are in\nneutron-lib too, in coming patch, when the gate is green\nagain, I will reopen the original patch that removes the\nextension from networking-sfc repository, see [2].\n\n[1]: https://review.opendev.org/q/topic:%22bug/2142034%22\n[2]: https://review.opendev.org/c/openstack/networking-sfc/+/553819\n\nRelated-Bug: #2142034\nRelated-Bug: #2150864\nChange-Id: I38ba09c2d5e79649e8693e626cceac8268586e46\nSigned-off-by: lajoskatona \u003clajos.katona@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/13da26ebc22953501ca0a7ff2c04e3bdd82b5e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/networking-sfc/commit/13da26ebc22953501ca0a7ff2c04e3bdd82b5e60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
