)]}'
{"id":"openstack%2Fneutron-dynamic-routing~808024","triplet_id":"openstack%2Fneutron-dynamic-routing~master~I5d773851136759117b048660d34209a601054a69","project":"openstack/neutron-dynamic-routing","branch":"master","topic":"payload_callback","hashtags":[],"change_id":"I5d773851136759117b048660d34209a601054a69","subject":"Use payload style callback","status":"MERGED","created":"2021-09-09 09:06:51.000000000","updated":"2021-09-15 09:38:47.000000000","submitted":"2021-09-15 09:37:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808024-payload_callback","meta_rev_id":"4870da2aa071a797682f17cd343c53913ca09d98","_number":808024,"virtual_id_number":808024,"owner":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 09:37:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"date":"2021-09-15 08:26:33.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2021-09-14 09:43:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-13 13:22:02.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2021-09-15 08:58:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-09 10:03:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-13 12:31:00.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2021-09-13 14:15:02.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-09-13 14:15:02.000000000","updated_by":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-09-15 07:13:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"bbc36b2d227a22923e8e015b3e4f169ce1d357ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-09 09:06:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afd3334d241256e81348a8a5bb50037c7958a2ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 10:03:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c2dcbb8f730458ab66dd9d0b400764c : FAILURE in 6m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f0315a7550446a383a6c01abe6a3ced : FAILURE in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1afab51cfa08420f8cd201cd954fd643 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/949a23b9457c4412b240657679b46ce1 : FAILURE in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/290fec437e06441baf16b5a3e0801210 : FAILURE in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e16e1d1025804117a37cc34072fc5e79 : FAILURE in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10b5b04617a4edabe679cf7320b6359 : SUCCESS in 7m 42s\n- neutron-dynamic-routing-functional https://zuul.opendev.org/t/openstack/build/d421092f0bda4a35afa10a6a17d0cce2 : SUCCESS in 17m 28s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/e7296b3c492e40b18c4cd734019fed4a : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c2bbae1737d5a5d662e3c2f4184a4d9c7b8d57f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-09 14:42:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ee00227a5570c33691ae96d184a9d1adadad91f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 15:28:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9832c7d27454e4d8bdabc349d615b6a : FAILURE in 8m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3c4334a9f774a5f96f4eb98e579565b : FAILURE in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a0adacdded84ec69cfbb8521c8cae79 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/496f8b162d4547d0b7df56f0ff99fa33 : FAILURE in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6968d76fba094e69963358d43e8d18f6 : FAILURE in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4cc274255a04456ab0d6f784bb99268 : FAILURE in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0fdb20f2c59440c9574c22dc1eaeb77 : SUCCESS in 8m 08s\n- neutron-dynamic-routing-functional https://zuul.opendev.org/t/openstack/build/d533367b71ce4675a8ddadc731d81011 : SUCCESS in 24m 39s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/fa7ec43689144ef1b102485f05481350 : SUCCESS in 41m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9317b3c1621c356b7dd893e1125304d13b737078","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-13 12:14:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e169625725d9bac9499a98637c67f0a90a8af590","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"date":"2021-09-13 12:31:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"eefd2855dbe375c282db098e21de84a388af11d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-13 13:18:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbccd0b0c0004a6f8424ab0c0ef26396 : SUCCESS in 6m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20b27ac4df8c47568b81c468584d064f : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a03379b1de5c4c36bdf63237bcb48cb0 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/023dbc080df14bd0a9a288f2f3ccce0c : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7efb8e6f4f3840d39c379bf4c721b069 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92241cd90eaf4a0fa23e8c96f839ec4f : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f40e32ed6e244b28b0dff4d61de1d50 : SUCCESS in 6m 35s\n- neutron-dynamic-routing-functional https://zuul.opendev.org/t/openstack/build/6db48a85401749d699c39be6c5ecc1e1 : SUCCESS in 21m 54s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/3bfd603f624f4522ac14ea92c1e3d9a9 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":4},{"id":"5e95ee4d8494a44dc0050bd7ee6b7b1fceb7ead0","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-13 13:22:02.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b496c23c7189a0a3d5926a7f8658faa54104881e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-14 09:43:10.000000000","message":"Patch Set 4: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":4},{"id":"1485ff2078af3ab571b050743b44a35d27d08120","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-09-15 07:13:15.000000000","message":"Patch Set 4: Code-Review-1\n\nWill this need special handling during upgrades? I think at least a reno would be needed.","accounts_in_message":[],"_revision_number":4},{"id":"2d268ccdfd28f8e3d1f9c4275d613861d36f9144","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-15 07:16:56.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review-1\n\u003e \n\u003e Will this need special handling during upgrades? I think at least a reno would be needed.\n\nI don\u0027t see any upgrade impact in this patch, it just changes which format to use for internal notifications. What you think should reno say?","accounts_in_message":[],"_revision_number":4},{"id":"1b5cbcf8a72cee9c45cc73d976f46c5ce890651d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-09-15 07:20:47.000000000","message":"Patch Set 4:\n\nWell this changes how the bgp-dr-agent talks with the neutron server, doesn\u0027t it? Or is this callback only internal to the agent?","accounts_in_message":[],"_revision_number":4},{"id":"9615ecb4c30cf01ca3749d8dd3d7f2034d44db34","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-15 07:44:08.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e Well this changes how the bgp-dr-agent talks with the neutron server, doesn\u0027t it? Or is this callback only internal to the agent?\n\nThis event (BGP_SPEAKER AFTER_CREATE ) is server internal as I see, but Yes doesn\u0027t affect server-agent communication","accounts_in_message":[],"_revision_number":4},{"id":"8c014ae73505ef6821fbec6b5aa1d46d3128f91b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2021-09-15 07:56:51.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e \u003e Patch Set 4:\n\u003e \u003e \n\u003e \u003e Well this changes how the bgp-dr-agent talks with the neutron server, doesn\u0027t it? Or is this callback only internal to the agent?\n\u003e \n\u003e This event (BGP_SPEAKER AFTER_CREATE ) is server internal as I see, but Yes doesn\u0027t affect server-agent communication\n\nRegistry notifications are always within same process, there is no RPC interface underneath.","accounts_in_message":[],"_revision_number":4},{"id":"e001edeef4d11937274642f6d912ad002446bfa4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-09-15 08:26:33.000000000","message":"Patch Set 4: -Code-Review\n\no.k., nevermind then, go ahead","accounts_in_message":[],"_revision_number":4},{"id":"4c1433ef50d2251ae3a59d5b3332e13ee20b84c9","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-09-15 08:58:47.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6a8c9cb90843b8e2874c8b4bc1224d1680cd05b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 08:59:22.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0ffdb30a625be868d9296c482124ace7b3364d5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:37:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd92b955294e4490a660c3c088acb130 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a14503b1c1c45fe8fe69cd67b4a61f8 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6b423032a4d45a6a461a7a20e0285f1 : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4538efe677d4d80a72f4766949ed205 : SUCCESS in 10m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37a278b446f248ceb35b78b1a9961d26 : SUCCESS in 5m 49s\n- neutron-dynamic-routing-functional https://zuul.opendev.org/t/openstack/build/5c703ac03f6140af94161908d97f786b : SUCCESS in 27m 25s\n- neutron-tempest-plugin-dynamic-routing https://zuul.opendev.org/t/openstack/build/73cbe47b26f943108f22d2066e8003d1 : SUCCESS in 36m 08s","accounts_in_message":[],"_revision_number":4},{"id":"b6148a88c8e05fd49f0ae89990d798931b98ca84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:37:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4870da2aa071a797682f17cd343c53913ca09d98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 09:38:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980cffa21e7e4ce5a7649fc23b94b0eb : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cae03eaa496b149c8655fe4e76a5538e51b106e3","revisions":{"fab77f993e76f1cda0969995b8cebbed726aa659":{"kind":"REWORK","_number":1,"created":"2021-09-09 09:06:51.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/24/808024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-dynamic-routing","ref":"refs/changes/24/808024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/1"}}},"commit":{"parents":[{"commit":"891ca86212f0f9fdce669d44b65daa9363d1b847","subject":"use payloads for ROUTER_INTERFACE events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/891ca86212f0f9fdce669d44b65daa9363d1b847"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-09 09:04:50.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-09 09:06:49.000000000","tz":180},"subject":"Use payload style callback","message":"Use payload style callback\n\nNeutron-lib registry.notify will be removed\n\nChange-Id: I5d773851136759117b048660d34209a601054a69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/fab77f993e76f1cda0969995b8cebbed726aa659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/fab77f993e76f1cda0969995b8cebbed726aa659"}]},"branch":"refs/heads/master"},"966a82389d33ae41d50295bc60394f3a3ada85a4":{"kind":"REWORK","_number":2,"created":"2021-09-09 14:42:42.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/24/808024/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-dynamic-routing","ref":"refs/changes/24/808024/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/2"}}},"commit":{"parents":[{"commit":"891ca86212f0f9fdce669d44b65daa9363d1b847","subject":"use payloads for ROUTER_INTERFACE events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/891ca86212f0f9fdce669d44b65daa9363d1b847"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-09 09:04:50.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-09 14:42:10.000000000","tz":180},"subject":"Use payload style callback","message":"Use payload style callback\n\nNeutron-lib registry.notify will be removed\n\nChange-Id: I5d773851136759117b048660d34209a601054a69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/966a82389d33ae41d50295bc60394f3a3ada85a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/966a82389d33ae41d50295bc60394f3a3ada85a4"}]},"branch":"refs/heads/master"},"203c0a961bf8f17b7e61ff5f366530a7ee7db8b1":{"kind":"REWORK","_number":3,"created":"2021-09-13 12:14:27.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/24/808024/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-dynamic-routing","ref":"refs/changes/24/808024/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/3"}}},"commit":{"parents":[{"commit":"891ca86212f0f9fdce669d44b65daa9363d1b847","subject":"use payloads for ROUTER_INTERFACE events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/891ca86212f0f9fdce669d44b65daa9363d1b847"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-09 09:04:50.000000000","tz":180},"committer":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-13 12:14:14.000000000","tz":180},"subject":"Use payload style callback","message":"Use payload style callback\n\nNeutron-lib registry.notify will be removed\n\nChange-Id: I5d773851136759117b048660d34209a601054a69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/203c0a961bf8f17b7e61ff5f366530a7ee7db8b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/203c0a961bf8f17b7e61ff5f366530a7ee7db8b1"}]},"branch":"refs/heads/master"},"cae03eaa496b149c8655fe4e76a5538e51b106e3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-09-13 12:31:13.000000000","uploader":{"_account_id":32667,"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","username":"isabek"},"ref":"refs/changes/24/808024/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-dynamic-routing","ref":"refs/changes/24/808024/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-dynamic-routing refs/changes/24/808024/4"}}},"commit":{"parents":[{"commit":"630678b5f9cdaaef0918d018a6f3b4f1c58ce533","subject":"Payload changes for FLOATING_IP events","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/630678b5f9cdaaef0918d018a6f3b4f1c58ce533"}]}],"author":{"name":"Nurmatov Mamatisa","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-09 09:04:50.000000000","tz":180},"committer":{"name":"Mamatisa Nurmatov","email":"nurmatov.mamatisa@huawei.com","date":"2021-09-13 12:31:13.000000000","tz":0},"subject":"Use payload style callback","message":"Use payload style callback\n\nNeutron-lib registry.notify will be removed\n\nChange-Id: I5d773851136759117b048660d34209a601054a69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/cae03eaa496b149c8655fe4e76a5538e51b106e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-dynamic-routing/commit/cae03eaa496b149c8655fe4e76a5538e51b106e3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
