)]}'
{"id":"openstack%2Fneutron-interconnection~639735","triplet_id":"openstack%2Fneutron-interconnection~master~Ic1f1690423f2a61a04494259d29b85815f6826d4","project":"openstack/neutron-interconnection","branch":"master","topic":"plugin","hashtags":[],"change_id":"Ic1f1690423f2a61a04494259d29b85815f6826d4","subject":"Implements \u0027neutron-interconnection\u0027 plugin","status":"ABANDONED","created":"2019-02-27 16:29:32.000000000","updated":"2019-11-15 20:08:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"829fd5fb08c9f022755b86dc9159342e3b4b4bd9","_number":639735,"virtual_id_number":639735,"owner":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":1,"date":"2019-05-27 16:03:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2019-06-20 13:50:51.000000000","_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-27 16:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-20 13:50:51.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"}],"messages":[{"id":"88dcd9e32b3db7ef28685f100ffe8a0386c89b71","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-27 16:29:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc093dad8d89183a0c4dc10bc9eeada3260becc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 16:30:01.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"1a09558294bf25c72a8c7d106799e8a7c9d4a84c","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-27 17:00:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f41488880283823b98953eedea11f9ace55beabc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 17:17:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/639735/2/check/openstack-tox-lower-constraints/933c824/ : SUCCESS in 8m 27s\n- openstack-tox-pep8 http://logs.openstack.org/35/639735/2/check/openstack-tox-pep8/17cb669/ : SUCCESS in 7m 11s\n- openstack-tox-py27 http://logs.openstack.org/35/639735/2/check/openstack-tox-py27/15f7ecf/ : SUCCESS in 6m 23s\n- openstack-tox-docs http://logs.openstack.org/35/639735/2/check/openstack-tox-docs/4fcf4ec/html/ : SUCCESS in 4m 45s\n- openstack-tox-py35 http://logs.openstack.org/35/639735/2/check/openstack-tox-py35/50720c3/ : SUCCESS in 7m 10s\n- openstack-tox-py36 http://logs.openstack.org/35/639735/2/check/openstack-tox-py36/3a202c1/ : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":2},{"id":"211ef87b36ad59245b66db63b9786932c05c6c36","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-04-01 09:59:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1c53a74ab1385d9c539cb32180e3dc9179869267","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 10:11:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/639735/3/check/openstack-tox-lower-constraints/c4b1f3c/ : FAILURE in 7m 14s\n- openstack-tox-pep8 http://logs.openstack.org/35/639735/3/check/openstack-tox-pep8/4e8aa67/ : SUCCESS in 6m 40s\n- openstack-tox-py27 http://logs.openstack.org/35/639735/3/check/openstack-tox-py27/1e87de0/ : FAILURE in 6m 39s\n- openstack-tox-docs http://logs.openstack.org/35/639735/3/check/openstack-tox-docs/c089975/html/ : SUCCESS in 5m 14s\n- openstack-tox-py35 http://logs.openstack.org/35/639735/3/check/openstack-tox-py35/f216cf4/ : FAILURE in 6m 39s\n- openstack-tox-py36 http://logs.openstack.org/35/639735/3/check/openstack-tox-py36/01d4c8d/ : FAILURE in 6m 43s","accounts_in_message":[],"_revision_number":3},{"id":"c57489d477050b288cda84f2c8843cb60f373942","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-04-02 09:28:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3c6fcae39ed8b55908027be27536504d82faa702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 09:37:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/639735/4/check/openstack-tox-lower-constraints/91dc6ab/ : SUCCESS in 6m 21s\n- openstack-tox-pep8 http://logs.openstack.org/35/639735/4/check/openstack-tox-pep8/54205ac/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/35/639735/4/check/openstack-tox-py27/fd9045d/ : SUCCESS in 6m 21s\n- openstack-tox-docs http://logs.openstack.org/35/639735/4/check/openstack-tox-docs/7e90609/html/ : SUCCESS in 5m 28s\n- openstack-tox-py35 http://logs.openstack.org/35/639735/4/check/openstack-tox-py35/b11fe7c/ : SUCCESS in 7m 07s\n- openstack-tox-py36 http://logs.openstack.org/35/639735/4/check/openstack-tox-py36/6720189/ : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":4},{"id":"f4194560779f4892b9f5444e467e91f52994499c","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-04-04 15:37:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"11e4b7446600466738c498cd18e5870702e5e815","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 16:02:42.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/639735/5/check/openstack-tox-lower-constraints/60dce6f/ : SUCCESS in 11m 02s\n- openstack-tox-pep8 http://logs.openstack.org/35/639735/5/check/openstack-tox-pep8/51c775d/ : SUCCESS in 6m 09s\n- openstack-tox-py27 http://logs.openstack.org/35/639735/5/check/openstack-tox-py27/702d1ca/ : SUCCESS in 6m 56s\n- openstack-tox-docs http://logs.openstack.org/35/639735/5/check/openstack-tox-docs/e66029a/html/ : SUCCESS in 5m 30s\n- openstack-tox-py35 http://logs.openstack.org/35/639735/5/check/openstack-tox-py35/b50ef77/ : SUCCESS in 9m 51s\n- openstack-tox-py36 http://logs.openstack.org/35/639735/5/check/openstack-tox-py36/8a63699/ : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":5},{"id":"a2b699c00b98129155c2d1f1555cd05f3988de76","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-05-27 15:30:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"78caa8b856a71c11c57ca93828555f90f62b73af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 15:49:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/639735/6/check/openstack-tox-lower-constraints/a07321e/ : SUCCESS in 9m 33s\n- openstack-tox-pep8 http://logs.openstack.org/35/639735/6/check/openstack-tox-pep8/e21f2ae/ : FAILURE in 8m 02s\n- openstack-tox-py27 http://logs.openstack.org/35/639735/6/check/openstack-tox-py27/b127713/ : SUCCESS in 7m 03s\n- openstack-tox-docs http://logs.openstack.org/35/639735/6/check/openstack-tox-docs/9d47ae1/html/ : SUCCESS in 7m 00s\n- openstack-tox-py35 http://logs.openstack.org/35/639735/6/check/openstack-tox-py35/0896e15/ : SUCCESS in 7m 03s\n- openstack-tox-py36 http://logs.openstack.org/35/639735/6/check/openstack-tox-py36/b9169aa/ : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":6},{"id":"143dece27f316af2be3f4fd753a3c127ae97353f","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-05-27 15:51:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1b10bcbfb6196205fd9d1b873eed68fcdb349830","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 16:03:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/35/639735/7/check/openstack-tox-lower-constraints/3ad3d17/ : SUCCESS in 5m 40s\n- openstack-tox-pep8 http://logs.openstack.org/35/639735/7/check/openstack-tox-pep8/8611bb9/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/35/639735/7/check/openstack-tox-py27/8760238/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/35/639735/7/check/openstack-tox-docs/4b62689/html/ : SUCCESS in 4m 18s\n- openstack-tox-py35 http://logs.openstack.org/35/639735/7/check/openstack-tox-py35/d8c1f43/ : SUCCESS in 6m 40s\n- openstack-tox-py36 http://logs.openstack.org/35/639735/7/check/openstack-tox-py36/8980cdc/ : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":7},{"id":"0151578f674b0fbcb122c817b967c4dbaae01998","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-15 20:08:34.000000000","message":"Abandoned\n\nThe neutron-interconnection repo is being retired.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"249dd1453a96b3411d308556f0da0975c45a4ab1","revisions":{"49739ea1790787f32873325ad5835fe75b068d0f":{"kind":"REWORK","_number":1,"created":"2019-02-27 16:29:32.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/35/639735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-interconnection","ref":"refs/changes/35/639735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/1"}}},"commit":{"parents":[{"commit":"b25d9958de8d845a94bcd231bf10adcb60e176c1","subject":"Add interconnection state scheduler worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/b25d9958de8d845a94bcd231bf10adcb60e176c1"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-27 10:28:47.000000000","tz":60},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-27 15:49:07.000000000","tz":60},"subject":"Implements \u0027neutron-interconnection\u0027 plugin","message":"Implements \u0027neutron-interconnection\u0027 plugin\n\n* Defines \u0027neutron-interconnection\u0027 plugin.\n* Adds driver support with dummy test driver.\n* Adds corresponding unit tests.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ic1f1690423f2a61a04494259d29b85815f6826d4\nDepends-On: https://review.openstack.org/631254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/49739ea1790787f32873325ad5835fe75b068d0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/49739ea1790787f32873325ad5835fe75b068d0f"}]},"branch":"refs/heads/master"},"504034414bb98296612731259e50f39fe52d0e72":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2019-02-27 17:00:23.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/35/639735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-interconnection","ref":"refs/changes/35/639735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/2"}}},"commit":{"parents":[{"commit":"aca715a770c57af2166837b1ff38d57d307aede8","subject":"Implements interconnection API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/aca715a770c57af2166837b1ff38d57d307aede8"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-27 10:28:47.000000000","tz":60},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-27 16:59:56.000000000","tz":60},"subject":"Implements \u0027neutron-interconnection\u0027 plugin","message":"Implements \u0027neutron-interconnection\u0027 plugin\n\n* Defines \u0027neutron-interconnection\u0027 plugin.\n* Adds driver support with dummy test driver.\n* Adds corresponding unit tests.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ic1f1690423f2a61a04494259d29b85815f6826d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/504034414bb98296612731259e50f39fe52d0e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/504034414bb98296612731259e50f39fe52d0e72"}]},"branch":"refs/heads/master"},"e265a0c90b32a2dd96082821f9d724904c90aa33":{"kind":"REWORK","_number":3,"created":"2019-04-01 09:59:39.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/35/639735/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-interconnection","ref":"refs/changes/35/639735/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/3"}}},"commit":{"parents":[{"commit":"fc7891a8a5d93981bec15715263a9ed3195a8e6f","subject":"Implements interconnection API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/fc7891a8a5d93981bec15715263a9ed3195a8e6f"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-27 10:28:47.000000000","tz":60},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-04-01 09:58:40.000000000","tz":120},"subject":"Implements \u0027neutron-interconnection\u0027 plugin","message":"Implements \u0027neutron-interconnection\u0027 plugin\n\n* Defines \u0027neutron-interconnection\u0027 plugin.\n* Adds driver support with dummy test driver.\n* Adds corresponding unit tests.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ic1f1690423f2a61a04494259d29b85815f6826d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/e265a0c90b32a2dd96082821f9d724904c90aa33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/e265a0c90b32a2dd96082821f9d724904c90aa33"}]},"branch":"refs/heads/master"},"8501869e7a7f7de5c13ddd321ae8152826b7713e":{"kind":"REWORK","_number":4,"created":"2019-04-02 09:28:13.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/35/639735/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-interconnection","ref":"refs/changes/35/639735/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/4"}}},"commit":{"parents":[{"commit":"32928283f574d8e2659b9e9dbdd1c2dd5487883d","subject":"Implements interconnection API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/32928283f574d8e2659b9e9dbdd1c2dd5487883d"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-27 10:28:47.000000000","tz":60},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-04-02 09:27:56.000000000","tz":120},"subject":"Implements \u0027neutron-interconnection\u0027 plugin","message":"Implements \u0027neutron-interconnection\u0027 plugin\n\n* Defines \u0027neutron-interconnection\u0027 plugin.\n* Adds driver support with dummy test driver.\n* Adds corresponding unit tests.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ic1f1690423f2a61a04494259d29b85815f6826d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/8501869e7a7f7de5c13ddd321ae8152826b7713e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/8501869e7a7f7de5c13ddd321ae8152826b7713e"}]},"branch":"refs/heads/master"},"fe8433194fb39eaf1d78a4d1431abce3e08f39b2":{"kind":"REWORK","_number":5,"created":"2019-04-04 15:37:18.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/35/639735/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-interconnection","ref":"refs/changes/35/639735/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/5"}}},"commit":{"parents":[{"commit":"32928283f574d8e2659b9e9dbdd1c2dd5487883d","subject":"Implements interconnection API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/32928283f574d8e2659b9e9dbdd1c2dd5487883d"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-27 10:28:47.000000000","tz":60},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-04-04 15:37:08.000000000","tz":120},"subject":"Implements \u0027neutron-interconnection\u0027 plugin","message":"Implements \u0027neutron-interconnection\u0027 plugin\n\n* Defines \u0027neutron-interconnection\u0027 plugin.\n* Adds driver support with dummy test driver.\n* Adds corresponding unit tests.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ic1f1690423f2a61a04494259d29b85815f6826d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/fe8433194fb39eaf1d78a4d1431abce3e08f39b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/fe8433194fb39eaf1d78a4d1431abce3e08f39b2"}]},"branch":"refs/heads/master"},"a1ea0642ff1bfd7b8a082fa95aa0fde2e71e23d0":{"kind":"REWORK","_number":6,"created":"2019-05-27 15:30:35.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/35/639735/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-interconnection","ref":"refs/changes/35/639735/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/6"}}},"commit":{"parents":[{"commit":"32928283f574d8e2659b9e9dbdd1c2dd5487883d","subject":"Implements interconnection API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/32928283f574d8e2659b9e9dbdd1c2dd5487883d"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-27 10:28:47.000000000","tz":60},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-05-27 15:31:24.000000000","tz":120},"subject":"Implements \u0027neutron-interconnection\u0027 plugin","message":"Implements \u0027neutron-interconnection\u0027 plugin\n\n* Defines \u0027neutron-interconnection\u0027 plugin.\n* Adds driver support with dummy test driver.\n* Adds corresponding unit tests.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ic1f1690423f2a61a04494259d29b85815f6826d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/a1ea0642ff1bfd7b8a082fa95aa0fde2e71e23d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/a1ea0642ff1bfd7b8a082fa95aa0fde2e71e23d0"}]},"branch":"refs/heads/master"},"249dd1453a96b3411d308556f0da0975c45a4ab1":{"kind":"REWORK","_number":7,"created":"2019-05-27 15:51:08.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/35/639735/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-interconnection","ref":"refs/changes/35/639735/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-interconnection refs/changes/35/639735/7"}}},"commit":{"parents":[{"commit":"32928283f574d8e2659b9e9dbdd1c2dd5487883d","subject":"Implements interconnection API extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/32928283f574d8e2659b9e9dbdd1c2dd5487883d"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-27 10:28:47.000000000","tz":60},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-05-27 15:52:06.000000000","tz":120},"subject":"Implements \u0027neutron-interconnection\u0027 plugin","message":"Implements \u0027neutron-interconnection\u0027 plugin\n\n* Defines \u0027neutron-interconnection\u0027 plugin.\n* Adds driver support with dummy test driver.\n* Adds corresponding unit tests.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: Ic1f1690423f2a61a04494259d29b85815f6826d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/249dd1453a96b3411d308556f0da0975c45a4ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-interconnection/commit/249dd1453a96b3411d308556f0da0975c45a4ab1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
