)]}'
{"id":"openstack%2Fneutron-lib~1000127","triplet_id":"openstack%2Fneutron-lib~master~I32154a2af5d6f34961212f69bd81c955bf3ad6d9","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2026-08-13 19:25:37.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-08-13 15:18:23.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2026-08-13 07:56:18.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}}},"hashtags":[],"change_id":"I32154a2af5d6f34961212f69bd81c955bf3ad6d9","subject":"Fix DriverBase.is_loaded","status":"MERGED","created":"2026-08-07 17:01:19.000000000","updated":"2026-08-13 19:27:02.000000000","submitted":"2026-08-13 19:25:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1000127","meta_rev_id":"42ca010fbedffd7a2c07fd84fa95e6d0790da498","_number":1000127,"virtual_id_number":1000127,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-13 19:25:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-08-12 00:51:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-08-12 07:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2026-08-13 07:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2026-08-13 07:56:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-07 19:07:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-08-10 09:27:30.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2026-08-12 00:51:04.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-08-12 07:20:52.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2026-08-13 07:56:18.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"211a6cf39f4d4ab07a2206235bf6b9ac6d1079eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-07 17:01:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbdb8b9ae5d5c0dfcb8c2401cfcfdcf03cdde6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-07 19:07:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3064603fb47d4e2092f7bb9b25ca7d76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3eb64394e54903880dc12a414aa2fa : SUCCESS in 8m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c9abbc40baf4d8281efa91c4f544e07 : FAILURE in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55f9b1f7bd71409d83c78a9a392616d3 : FAILURE in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/055121a41b184e2a82ced6d8dbc2c122 : FAILURE in 4m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cf62c0160a2043ecbadc38ba0dcaa3b9 : FAILURE in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ce6addbaac6e4226a0df8bec25570539 : FAILURE in 4m 56s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/65b416c232d4456f836022682cc11183 : FAILURE in 5m 20s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e923fe2846864a8f8c3da6f824b0a6b0 : SUCCESS in 1h 02m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8563c71f2874a02978fe4072fc95338 : SUCCESS in 4m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/af200d38eac640488f5d7b97ef80e2b4 : SUCCESS in 1h 01m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2144e9253b7e4cb09072fa224c80b7eb : SUCCESS in 2h 04m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/014a1166d3f14d60a86bd723e919494a : SUCCESS in 30m 37s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/94e7ff403f20499ca804f76f82c2c220 : SUCCESS in 21m 48s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/dea94016e29e48d9b4289215b6940833 : FAILURE in 6m 28s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d7e27177f76d43d49a7a09b543886d83 : SUCCESS in 1h 13m 47s\n\nWarning:\n  Comments left for invalid file neutron_lib/tests/unit/services/qos/test_base.py","accounts_in_message":[],"_revision_number":1},{"id":"8a84ad1214c435074ebaf9875e891fbdde1d9829","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-10 00:49:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"a75a9b77e29222d59e153dbb93bd150b63764fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 02:36:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b138d88f8ec4d45a2f30f35f02f71dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f6354969d7c4a1195ebfab257855212 : SUCCESS in 6m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f3a85cea17c4db38cdd4ba4dae9b950 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f71363a272b14f9f80e89607b953e3cd : SUCCESS in 3m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8ead0d9ec254194a8568a875b0bdf94 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3f97066cf9034aa3b1c8e7f9fef08600 : SUCCESS in 2m 53s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d8596a5b9f514cc8b0c47c1bcc5a84d0 : SUCCESS in 5m 55s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/38347298b04f483895caf874219c63e2 : FAILURE in 4m 56s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23269a13bff24feba3884ae0d7d25cd3 : SUCCESS in 1h 08m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29ab640c19ea44b583e7023c0ba84bda : SUCCESS in 2m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bd837578387a426ba970a472e25f4623 : SUCCESS in 1h 06m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/93ffee0cdba746bba00de139f2391de4 : SUCCESS in 1h 40m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/327b5068d5654118836317cdfe29fc2d : SUCCESS in 30m 24s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/36245f8e47a448f2b7d484e0bcfa289c : SUCCESS in 19m 42s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/2ce74d315520455ba258d20acf21bd9c : SUCCESS in 5m 38s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/13d2ffa4d8154d14865259e25cd210d8 : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":2},{"id":"4df1a210a75f7bb4d0c8cf6ada47b8bac47cfe39","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-10 09:27:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a7697119d8ccf9325d70df819761ca51ca30276","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-11 02:27:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5535f96475f202724316783c8ff83130c387bc51","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-08-12 00:51:04.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a16de3198e29dcb08b1c0407898011309752e7f6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2026-08-12 07:20:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"51fb55d126944b9b68943055ca39a87b05f7f16c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-08-13 07:56:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c37ca62d498e736423595e887c30034faa01025","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 07:56:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5a915ce6ccd0f67258d251074f7a6430bfcf883","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 10:50:19.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cba9e701cad0424a998591bf6d96bf83\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ca465892d9458b97d483aec1740811 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eb29ae5ba9c4647957afaa5b1b8fd55 : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9e340b895794d3787faac84824c4153 : SUCCESS in 2m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1ac8b4742f1e47dd9734c2809e583dd8 : SUCCESS in 2m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/15a2a83460c54fb2bb5bad4e0cd8183b : SUCCESS in 6m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e9eb78daf4342948a75bd1261e64fb1 : FAILURE in 1h 16m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbf668172b8b49bd863e3ef43423181a : SUCCESS in 4m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a3bf2496d34242ca83fa226692d3c383 : SUCCESS in 1h 17m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6e0083de3a604c18900769aeaf5beabb : SUCCESS in 2h 40m 47s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/4568cb52be104a24a06f3988149a6cb3 : SUCCESS in 1h 36m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7aff935e113351b12380d62044c547dc9e2c773b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-08-13 15:18:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"214dd9626172e0f51fb5fb7a7d1936f138898673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:26:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/614749b7b91a4108a571d1178355858e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a13dba19244aba8448e27c409aa805 : SUCCESS in 8m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ef756b3f6ef4a918c15ae9efe2a4e4b : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f03a132f47445d6a3fa19524f4a6db7 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c5828f7e2ce429eb8423ba73d577f93 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc73405bfa6b467395e6335eff5245bf : SUCCESS in 5m 58s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/fe211b88458b4689b1365fa8a39bed2c : SUCCESS in 7m 59s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/d1d9eae174814e7dbbb1c4783d633d7c : FAILURE in 3m 22s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0eaa6d0bd5b401c8b875fc34aabd01d : SUCCESS in 1h 38m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f96eb760036b4dcdb55bf2953aafddc8 : SUCCESS in 5m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4e25671eff9942b9944d7a5dca62a97a : SUCCESS in 1h 00m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9b3ef77253414536a263a3b49b381134 : SUCCESS in 2h 03m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/24cee9004feb4800bd209d0803093410 : SUCCESS in 30m 40s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/5975d7396c3a4e8c8d267d32c3f90bb7 : SUCCESS in 21m 14s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/e1b46a9a14ab4d3cb373c27622a2af6a : SUCCESS in 5m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6e2ca455ca7d4a8591f66497a75e01a4 : SUCCESS in 1h 40m 13s","accounts_in_message":[],"_revision_number":2},{"id":"2c2efefdbd9d7591308c99e9bd7ae6a79b94dead","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 17:26:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c92c0a63ea11e496197457fd6e741eb96a92b651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:25:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcb48fc20023425dbd8efc94b7a6a401\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44c4a2eb548d47a68506dafcbf3939d7 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6968dd465b2042e08d41e50cbbb10d7f : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a6c9b5c2e524c9ea3fc9dcfe47b6c7e : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6061839279ca4fc88572292bd3b3645a : SUCCESS in 3m 09s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/24a422ee5dc34478bdd28fffe014c8d7 : SUCCESS in 6m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c13fbd04fb74eefb498ea139342b055 : SUCCESS in 40m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26e26ef4c3394cd29e70bef58554e174 : SUCCESS in 2m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dc5db15dd5854266b8c4eed67bb37070 : SUCCESS in 1h 06m 11s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c85a697d1d244b48b91688d61ebc87bb : SUCCESS in 1h 57m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/23d1f66390ed49cead41e5c8a87a0424 : SUCCESS in 1h 31m 28s","accounts_in_message":[],"_revision_number":2},{"id":"439c1f11bff4b5a88d16b8e4cf56447ee4f4ca6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:25:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"42ca010fbedffd7a2c07fd84fa95e6d0790da498","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 19:27:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45139f6a69334bdc81c02880e06cb367\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857146d2196549679a4fe904bad3dbbd : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98e77d851493e731eaac7af9407e414da5d3d95b","revisions":{"4979879631440418cd439e313ea450b6f93768bf":{"kind":"REWORK","_number":1,"created":"2026-08-07 17:01:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/1000127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/27/1000127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/1000127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/1000127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/1000127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/27/1000127/1"}}},"commit":{"parents":[{"commit":"92b5fb83df3d481d27bd2244d34cd3839c6ba625","subject":"Set neutron-lib as dependency for hacking pre-commit hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/92b5fb83df3d481d27bd2244d34cd3839c6ba625"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-07 17:00:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-07 17:00:52.000000000","tz":540},"subject":"Fix DriverBase.is_loaded","message":"Fix DriverBase.is_loaded\n\nIt should be a property, rather than a bare method.\n\nChange-Id: I32154a2af5d6f34961212f69bd81c955bf3ad6d9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4979879631440418cd439e313ea450b6f93768bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4979879631440418cd439e313ea450b6f93768bf"}]},"branch":"refs/heads/master"},"98e77d851493e731eaac7af9407e414da5d3d95b":{"kind":"REWORK","_number":2,"created":"2026-08-10 00:49:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/1000127/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/27/1000127/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/1000127/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/1000127/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/1000127/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/27/1000127/2"}}},"commit":{"parents":[{"commit":"92b5fb83df3d481d27bd2244d34cd3839c6ba625","subject":"Set neutron-lib as dependency for hacking pre-commit hook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/92b5fb83df3d481d27bd2244d34cd3839c6ba625"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-07 17:00:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-08-10 00:49:02.000000000","tz":540},"subject":"Fix DriverBase.is_loaded","message":"Fix DriverBase.is_loaded\n\nIt should be a property, rather than a bare method.\n\nChange-Id: I32154a2af5d6f34961212f69bd81c955bf3ad6d9\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/98e77d851493e731eaac7af9407e414da5d3d95b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/98e77d851493e731eaac7af9407e414da5d3d95b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
