)]}'
{"id":"openstack%2Fneutron-lib~1003369","triplet_id":"openstack%2Fneutron-lib~master~Ia7994094fb9bffe0407ad1c8bb4f26031f46befc","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-02 09:50:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia7994094fb9bffe0407ad1c8bb4f26031f46befc","subject":"Fix typo","status":"MERGED","created":"2026-09-01 19:51:33.000000000","updated":"2026-09-02 09:51:38.000000000","submitted":"2026-09-02 09:50:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003369","meta_rev_id":"e076023f3e7d4b3c7f040cb7675b6a0bb370dd4f","_number":1003369,"virtual_id_number":1003369,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-02 09:50:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":2,"date":"2026-09-02 08:02:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2026-09-01 20:46:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":1,"date":"2026-09-02 08:02:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-01 20:46:06.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2026-09-01 21:34:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-02 08:02:13.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"b00a711e763f901077711c1a6630de206a715df0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:51:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"841982545683c9652124bd0ce82f3bfa3a26076f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:52:16.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"18e12c70f1594d5a591c0c8364a71c2031f50672","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-01 19:52:29.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"b13c13b0d2011180e90c4d2426e70203f9f43c50","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2026-09-01 20:46:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"14cdf2d04db85360363a4e13522637b8f22cfdf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 21:34:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f26c9a6f91734c99b4ec967d07583204\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dccae4fa7c545bba67f598dde74d25e : SUCCESS in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1548d4288e04f7795affbd358a7a9da : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b32727cbc54d38bb2388f18d5bec22 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3652b241ce5a477bb2fb00eff9a9208d : SUCCESS in 2m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd3e732497b9467d95487c018540caf4 : SUCCESS in 3m 10s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/67ccfa0357034a8c888bb64054e8a315 : SUCCESS in 5m 09s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4095e2b11ab9499da7764b91fe40962f : FAILURE in 5m 29s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ed5bd1e4fe14c8bb7749e7b0e0f22c4 : SUCCESS in 59m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80f7949fcebd4507ac82fdd53cef04e6 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1b22ae9e55384d8ab4019d054c7f7999 : SUCCESS in 1h 02m 42s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/3951eecc5100466ebf0cb99ec2113d3c : SUCCESS in 1h 40m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a46b24744e004d83a82f1a777ecb6421 : SUCCESS in 32m 42s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/0fbcda0e752941f9b4be56d68c7e5820 : SUCCESS in 24m 20s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/a333b41ba6a749dc9bfcc539c0d1284e : SUCCESS in 5m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d81dd69929584599a72eb849752304fd : SUCCESS in 1h 35m 12s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/fips/enable-fips.yaml for job neutron-tempest-plugin-openvswitch.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/openstack-zuul-jobs/playbooks/multinode/enable-multinode.yaml for job neutron-tempest-plugin-openvswitch.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/devstack/playbooks/pre.yaml for job neutron-tempest-plugin-openvswitch.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/devstack-tempest.yaml for job neutron-tempest-plugin-openvswitch.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/openstack/tempest/playbooks/post-tempest.yaml for job neutron-tempest-plugin-openvswitch.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"4e32d33aaa8dcc910bea0c284762df9703572db1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-02 08:02:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1ee57fb087adc08b55bc0f85569d5c723c6edd23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 08:02:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6342f0c6f871396974f4cf55047f93581a80eaab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 09:50:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd0c4ed5dae2484899d572a4bac5558f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/220a3d3a66db42e6bb930813179c2f1f : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88c1fc172fdf4eabb2d15530c9eccb5a : SUCCESS in 6m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af1daf77b72941cc91ab1849eb32bb41 : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd11fa2ee9b14183a7fbad1c9d2c6aab : SUCCESS in 3m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b6f82531277b40d9a99df1a3d627f40f : SUCCESS in 7m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f921b3dee2b447218f3b86fe7f6a1dd6 : SUCCESS in 1h 42m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f89dbf1681054d069c7fc34b7c21b857 : SUCCESS in 2m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a6bd07cca2f04a1582d352a1a5d3e303 : SUCCESS in 58m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/39ebe89c1ddb457781151003d08fa2bd : SUCCESS in 1h 36m 37s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/2385868d63be4e7ea39d736047d8aab0 : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":2},{"id":"0403074af6bc8a13041e531ed73b91df5b06937a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 09:50:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e076023f3e7d4b3c7f040cb7675b6a0bb370dd4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 09:51:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b82de07f74b2409cb94aeafc63c38f75\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/152fb8b7d4ef4e26b31a4b72b39ed8ed : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"857cd21e779bccc6f334b853766f8998e9db7a16","revisions":{"63071067ae36f18276d75db1023ac3f642d9bd4e":{"kind":"REWORK","_number":1,"created":"2026-09-01 19:51:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/1003369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/1003369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/1003369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/1003369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/1003369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/1003369/1"}}},"commit":{"parents":[{"commit":"fd8dac25a29c362cb1ce70763cdb9fd81eea11eb","subject":"placement: Raise ``PlacementResourceProviderNotFound`` on delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fd8dac25a29c362cb1ce70763cdb9fd81eea11eb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:51:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:51:33.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ia7994094fb9bffe0407ad1c8bb4f26031f46befc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/63071067ae36f18276d75db1023ac3f642d9bd4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/63071067ae36f18276d75db1023ac3f642d9bd4e"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"857cd21e779bccc6f334b853766f8998e9db7a16":{"kind":"REWORK","_number":2,"created":"2026-09-01 19:52:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/69/1003369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/69/1003369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/1003369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/1003369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/69/1003369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/69/1003369/2"}}},"commit":{"parents":[{"commit":"fd8dac25a29c362cb1ce70763cdb9fd81eea11eb","subject":"placement: Raise ``PlacementResourceProviderNotFound`` on delete","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fd8dac25a29c362cb1ce70763cdb9fd81eea11eb"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:51:33.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-01 19:52:16.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ia7994094fb9bffe0407ad1c8bb4f26031f46befc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/857cd21e779bccc6f334b853766f8998e9db7a16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/857cd21e779bccc6f334b853766f8998e9db7a16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
