)]}'
{"id":"openstack%2Fneutron-lib~1007331","triplet_id":"openstack%2Fneutron-lib~master~I9ab3db4b2c7cd4e5ec21f5d1ae910e0cdce90009","project":"openstack/neutron-lib","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-28 11:22:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9ab3db4b2c7cd4e5ec21f5d1ae910e0cdce90009","subject":"Install packages to build wheel of lxml","status":"MERGED","created":"2026-09-25 00:37:58.000000000","updated":"2026-09-28 11:25:00.000000000","submitted":"2026-09-28 11:22:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1007331","meta_rev_id":"ccb55306453bc29fc2c71fceaa5695199aaaa6ba","_number":1007331,"virtual_id_number":1007331,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-28 11:22:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2026-09-25 09:22:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2026-09-28 09:12:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2026-09-28 09:12:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-25 02:46:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-25 09:22:06.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-09-28 09:12:00.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"c14f8313d3299e942aefefd5f19ab6e67ed09fbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-09-25 00:37:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dde16981269a4327baea2c7767b5bd01522f5b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-25 02:46:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f29a250bf4cf46f5900b8bf1a8a3604c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/93be2ace80fa43179db5647b67e43dba : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b5b1ee2f5e44f78e2ba47f85080e60 : SUCCESS in 9m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bda2312560554ca697de5e6f487cc438 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47479d6bf4144afaa69a20ff24c25857 : SUCCESS in 7m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76822f78521f41f0b4a5c1c7fcc7e800 : SUCCESS in 4m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/389ca079b804464ca7102e99a91f4155 : SUCCESS in 6m 52s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/59272ff7b6324abb8216b5ed53b0379f : SUCCESS in 13m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85104751d6d64de4950a33c6a8565b66 : SUCCESS in 46m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f0f5305a00b498bbc99751602c380ff : SUCCESS in 5m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c7755bd0d42e4a5ba017d4a8d2844379 : SUCCESS in 1h 16m 18s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/36ae75c392fd403ba3ea7042a9a6013b : SUCCESS in 1h 45m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/61e0b47a1fdb4de6bed88ddfced509a9 : SUCCESS in 32m 15s\n- openstack-tox-py313-with-neutron https://zuul.opendev.org/t/openstack/build/1294bffb87b34127a1c1deebe3262f60 : SUCCESS in 26m 09s\n- openstack-tox-py313-with-sqlalchemy-master https://zuul.opendev.org/t/openstack/build/7ea3ba0cbacd479a92b3a01973400843 : SUCCESS in 6m 06s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/023fd2b9755049fda1996ec617e237a4 : SUCCESS in 1h 48m 03s","accounts_in_message":[],"_revision_number":1},{"id":"5455b257f383f6468ab76169cea6c1eb3ad52fa8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2026-09-25 09:22:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4fc8647b0be82cf12fcaa1a8367ffc53d2afeb7","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2026-09-28 09:12:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3319ba6ce2e891b42a56d1f1c050190e2f508cf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 09:12:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b556a558d2adf0a895ed8a52e11ece082e400dd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:22:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/374f27fd603b493a9f58b6a2a0c85f5f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d90803bee16e4ccea8bf5ecf76649541 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c1694658c44fdf9a82492acc368311 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05161a8f240746a89b9a8bd02b43f7a1 : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd201d493a424735ae032ee9e7cbe695 : SUCCESS in 4m 13s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/983913d739d941b8be733f4dbb900dc0 : SUCCESS in 7m 03s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/acfc1e9728d641a59cc9e447347e04b2 : SUCCESS in 7m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b2f2e7f08c8413796679eec026e6e5b : SUCCESS in 51m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7058a17b529047d8bd4994d8f80f37e8 : SUCCESS in 6m 37s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d9310d0ea41a44b8ae007cccdada64f9 : SUCCESS in 1h 06m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f5ff6fde39774970ad33c1a38d7a709d : SUCCESS in 2h 08m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6bb0c0bc1d3745df8eeb0b5519625f34 : SUCCESS in 1h 51m 19s","accounts_in_message":[],"_revision_number":1},{"id":"3deeb0feaa3e7b724fc8cef8e5c0bc1e7769e279","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:22:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ccb55306453bc29fc2c71fceaa5695199aaaa6ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-28 11:25:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17b4bbb400b4409898bf13cfae43deea\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/b04ecc0111bf48518418e2f281cfff3a : SUCCESS in 1m 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6704601568004f8094369807528c7e2a : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5df0c4d0109c428cb872e6a26f5626da : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"703d59f2e0ecafee4c93c795bc243151f8f9ea5d","revisions":{"703d59f2e0ecafee4c93c795bc243151f8f9ea5d":{"kind":"REWORK","_number":1,"created":"2026-09-25 00:37:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/31/1007331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/31/1007331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/1007331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/1007331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/31/1007331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/31/1007331/1"}}},"commit":{"parents":[{"commit":"5a5725ba76fbf74bef32acab4fe0790671c41480","subject":"Merge \"db: drop SQLAlchemy 1.x compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5a5725ba76fbf74bef32acab4fe0790671c41480"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 00:36:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-09-25 00:37:39.000000000","tz":540},"subject":"Install packages to build wheel of lxml","message":"Install packages to build wheel of lxml\n\nThe package is an indirect dependency via os-ken -\u003e ncclient .\n\nChange-Id: I9ab3db4b2c7cd4e5ec21f5d1ae910e0cdce90009\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/703d59f2e0ecafee4c93c795bc243151f8f9ea5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/703d59f2e0ecafee4c93c795bc243151f8f9ea5d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
