)]}'
{"id":"openstack%2Fneutron-lib~577220","triplet_id":"openstack%2Fneutron-lib~master~I192d99673feba97a95af995923b266e2f8b58c6d","project":"openstack/neutron-lib","branch":"master","topic":"minimum-bandwidth-allocation-placement-api","hashtags":[],"change_id":"I192d99673feba97a95af995923b266e2f8b58c6d","subject":"Placement: utils","status":"MERGED","created":"2018-06-21 15:22:37.000000000","updated":"2018-08-22 21:05:12.000000000","submitted":"2018-08-22 21:05:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":42,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"577220-1534971912551-522f7acf","meta_rev_id":"3e84b1effec87a7328d9d8232604555ed412ddcd","_number":577220,"virtual_id_number":577220,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-08-22 07:58:18.000000000","_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2018-08-22 21:05:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":1,"date":"2018-08-21 17:08:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":2,"date":"2018-08-22 19:40:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2018-08-22 13:52:13.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2018-08-22 09:32:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2018-08-22 14:24:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":1,"date":"2018-08-22 19:40:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-26 15:41:10.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2018-07-10 14:02:07.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-07-30 13:42:21.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2018-08-13 07:34:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2018-08-21 17:08:09.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2018-08-22 09:32:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2018-08-22 13:52:13.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2018-08-22 14:24:19.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2018-08-22 19:40:59.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2018-08-22 21:05:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"245233899fa788db6b6aee0466fe8671af983aa8","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-06-21 15:22:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5cb6ecddd3edf798a99d23a22539a39da7f2da4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-21 17:06:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/1/check/build-openstack-sphinx-docs/3dc5ca7/html/ : SUCCESS in 6m 47s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/1/check/openstack-tox-pep8/8e6101e/ : SUCCESS in 6m 40s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/1/check/openstack-tox-py27/b48bdd1/ : SUCCESS in 6m 15s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/1/check/openstack-tox-py35/3b27be7/ : SUCCESS in 6m 48s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/1/check/legacy-tempest-dsvm-neutron-src/189d623/ : SUCCESS in 1h 33m 56s\n- openstack-tox-cover http://logs.openstack.org/20/577220/1/check/openstack-tox-cover/c94c418/cover/ : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/1/check/openstack-tox-lower-constraints/0eb4679/ : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2546575d31eb5211c56b1e54e33624692689c364","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-06-22 11:40:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7bb3eba2347aa3a44c76d292fa31ceeef20eb978","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-22 13:00:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/2/check/build-openstack-sphinx-docs/41804ca/html/ : SUCCESS in 4m 46s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/2/check/openstack-tox-pep8/1f61883/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/2/check/openstack-tox-py27/f562ec7/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/2/check/openstack-tox-py35/eb0b23d/ : SUCCESS in 4m 48s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/2/check/legacy-tempest-dsvm-neutron-src/c8bf0cd/ : SUCCESS in 1h 17m 37s\n- openstack-tox-cover http://logs.openstack.org/20/577220/2/check/openstack-tox-cover/a5a41f0/cover/ : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/2/check/openstack-tox-lower-constraints/50030db/ : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"046fded2ed0219a495288c533111d0c48551a6dd","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2018-06-26 15:41:10.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nI left two nits inline.\nCheers,\ngibi","accounts_in_message":[],"_revision_number":2},{"id":"fa46fc8ada1f77fc97a4ee4d2704940bdfb49a05","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-07-06 14:05:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2d5d1925095fe2acdb444f963cad00039493c302","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-07-06 14:08:01.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nFixed the nits, but moved the constants file to its own change.","accounts_in_message":[],"_revision_number":3},{"id":"aa8d8d96cc28b6d7ad52e5047180bb22d3ab5dc5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-06 15:32:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/3/check/build-openstack-sphinx-docs/e7e8481/html/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/3/check/openstack-tox-pep8/3523331/ : SUCCESS in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/3/check/openstack-tox-py27/4eed40e/ : SUCCESS in 4m 29s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/3/check/openstack-tox-py35/f6fb4d7/ : SUCCESS in 6m 15s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/3/check/legacy-tempest-dsvm-neutron-src/c54ef8e/ : SUCCESS in 1h 20m 34s\n- openstack-tox-cover http://logs.openstack.org/20/577220/3/check/openstack-tox-cover/dbca68f/cover/ : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/3/check/openstack-tox-lower-constraints/2ca07fd/ : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":3},{"id":"b89c194951c1cdad4ee3b32ec61bc3b27c569fbf","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-07-09 09:29:58.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"abd7eb10536a4ff10f35295cbf0df1c4b64dfd72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-09 11:04:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/4/check/build-openstack-sphinx-docs/1e1756e/html/ : SUCCESS in 6m 00s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/4/check/openstack-tox-pep8/36d1554/ : SUCCESS in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/4/check/openstack-tox-py27/e3c2c50/ : SUCCESS in 6m 09s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/4/check/openstack-tox-py35/1f0cdf7/ : SUCCESS in 4m 25s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/4/check/legacy-tempest-dsvm-neutron-src/4b3f63a/ : SUCCESS in 1h 31m 14s\n- openstack-tox-cover http://logs.openstack.org/20/577220/4/check/openstack-tox-cover/fbf8eec/cover/ : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/4/check/openstack-tox-lower-constraints/1bd3d17/ : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":4},{"id":"c09b8119c8b21a3423ddbb895902e7603a14a7cf","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-07-10 12:48:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"383a2771199ba616b98b531f059b2c3de3a8df0f","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-07-10 12:49:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"58df11692d6a34259d4d76ef7703a2b6c2d75e27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-10 14:01:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/6/check/build-openstack-sphinx-docs/f79805b/html/ : SUCCESS in 6m 36s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/6/check/openstack-tox-pep8/679e660/ : SUCCESS in 7m 05s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/6/check/openstack-tox-py27/120f2b6/ : SUCCESS in 7m 05s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/6/check/openstack-tox-py35/45a924e/ : FAILURE in 5m 14s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/6/check/legacy-tempest-dsvm-neutron-src/82e2e0d/ : FAILURE in 1h 06m 59s\n- openstack-tox-cover http://logs.openstack.org/20/577220/6/check/openstack-tox-cover/81f424e/ : FAILURE in 8m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/6/check/openstack-tox-lower-constraints/e859053/ : FAILURE in 6m 32s","accounts_in_message":[],"_revision_number":6},{"id":"7b1a28fc7eca312c4377c3cb993c4958f0b8f511","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-07-10 14:02:07.000000000","message":"Patch Set 6:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- legacy-tempest-dsvm-neutron-src: https://bugs.launchpad.net/bugs/1763070\n- openstack-tox-lower-constraints: unrecognized error\n- openstack-tox-py35: unrecognized error\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":6},{"id":"da3c1ee2775f837fb2b09c30474e0d23415b2770","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-07-12 13:55:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8d99629de4d692f0fdc2ca9855c71035df6c1b3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-12 15:46:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/7/check/build-openstack-sphinx-docs/2e1f2c0/html/ : SUCCESS in 7m 02s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/7/check/openstack-tox-pep8/ec388ce/ : SUCCESS in 6m 40s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/7/check/openstack-tox-py27/193fb39/ : SUCCESS in 6m 09s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/7/check/openstack-tox-py35/e41f925/ : SUCCESS in 6m 08s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/7/check/legacy-tempest-dsvm-neutron-src/92489be/ : SUCCESS in 1h 30m 46s\n- openstack-tox-cover http://logs.openstack.org/20/577220/7/check/openstack-tox-cover/be4b2c6/cover/ : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/7/check/openstack-tox-lower-constraints/6b63e27/ : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":7},{"id":"6631b4fa5493e9998f91a34c5a4049382889adb7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-07-19 09:58:43.000000000","message":"Patch Set 7: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e53eccb300adc909498cbb2b912010dcf339b088","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-07-20 11:53:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e300d2abfa8578017a037d1931d160e0b2fc584c","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-07-20 11:56:55.000000000","message":"Patch Set 8:\n\n(6 comments)\n\nThanks for the review Slawek.","accounts_in_message":[],"_revision_number":8},{"id":"85f7fe9bb552fc26c85dac7bf88a46d9cd87b76c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-20 14:19:32.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/8/check/build-openstack-sphinx-docs/cd8c52e/html/ : SUCCESS in 6m 40s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/8/check/openstack-tox-pep8/dad712c/ : SUCCESS in 6m 14s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/8/check/openstack-tox-py27/0ac04a3/ : SUCCESS in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/8/check/openstack-tox-py35/d1b99c6/ : SUCCESS in 6m 58s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/8/check/legacy-tempest-dsvm-neutron-src/129145d/ : SUCCESS in 1h 26m 00s\n- openstack-tox-cover http://logs.openstack.org/20/577220/8/check/openstack-tox-cover/d0ac25a/cover/ : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/8/check/openstack-tox-lower-constraints/9af9838/ : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":8},{"id":"b909477b29a358fffb5b5ed0f954df92221dde09","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2018-07-30 13:42:21.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nOne nit within, which points to a potential bug.\n\nCan you point me to the discussion that led to the desire to have stable uuids. Not debating its validity, just curious about the process.","accounts_in_message":[],"_revision_number":8},{"id":"f2a98bb2b9ca80cca68a607d7958ae5fa780815e","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-01 10:53:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f4af9cddd4e9c0fc4c795c00bd82f933e0d5a3f1","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-01 11:27:08.000000000","message":"Patch Set 9:\n\n(1 comment)\n\n\u003e Can you point me to the discussion that led to the desire to have\n \u003e stable uuids. Not debating its validity, just curious about the\n \u003e process.\n\nI think a bit of it can be found in the neutron spec and its comments, but let me summarize here:\n\nOriginally we thought of using the UUID of the neutron agent (as in openstack network agent list) also for the agent RP UUID. However the agent UUID may change (it\u0027s rare though, eg. after openstack network agent delete and re-create). As soon as we have allocations on some of the RPs we can\u0027t just delete and re-create the RPs. We would need to re-parent RPs. That we wanted to avoid.\n\nIt also helps to keep the communication needs low if both a neutron agent and the server is able to deterministically generate RP UUIDs from inputs they already have (via v5 UUIDs). For example if neutron-server had generated an RP UUID we could have needed a new message to send it back to the agent too.","accounts_in_message":[],"_revision_number":9},{"id":"30b59f579e7c87c93a84a30db2f251e8c8c9ec2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-01 12:19:53.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/20/577220/9/check/requirements-check/a84477b/ : SUCCESS in 3m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/9/check/build-openstack-sphinx-docs/421184b/html/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/9/check/openstack-tox-pep8/46a0633/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/9/check/openstack-tox-py27/b55742d/ : SUCCESS in 4m 59s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/9/check/openstack-tox-py35/4c3cf6f/ : SUCCESS in 4m 40s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/9/check/legacy-tempest-dsvm-neutron-src/c9343ba/ : SUCCESS in 1h 24m 01s\n- openstack-tox-cover http://logs.openstack.org/20/577220/9/check/openstack-tox-cover/7f1bbf6/cover/ : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/9/check/openstack-tox-lower-constraints/6e32332/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":9},{"id":"48114c75dd9caa92e717a9c1ac83307902898169","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2018-08-06 12:45:38.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"6bdcb90bc5c73f2b0d6e0c1ecbdbfecdb775d464","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-06 13:01:58.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6f643d91ba6838d3677efe93bf78b3362a62245a","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-06 13:02:13.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"78f337fc1f31d8162f2c11f5743f0d1b27b1bf4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-06 14:34:14.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/20/577220/10/check/requirements-check/708f14c/ : SUCCESS in 3m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/10/check/build-openstack-sphinx-docs/52376f7/html/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/10/check/openstack-tox-pep8/a70015e/ : SUCCESS in 6m 16s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/10/check/openstack-tox-py27/98ddeea/ : SUCCESS in 5m 41s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/10/check/openstack-tox-py35/45b400c/ : SUCCESS in 5m 42s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/10/check/legacy-tempest-dsvm-neutron-src/0aa2dce/ : SUCCESS in 1h 20m 23s\n- openstack-tox-cover http://logs.openstack.org/20/577220/10/check/openstack-tox-cover/5af0d7c/cover/ : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/10/check/openstack-tox-lower-constraints/b95523d/ : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":10},{"id":"e82ed88ef95a372f89c84fc5a9693d5c0cb98254","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2018-08-07 10:17:54.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"af1152c9b19056d5bd4b4ca062ae0f51c5ae4949","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-08-07 22:33:50.000000000","message":"Patch Set 10: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d4770454283ac8f124ede37756aaeefe875bd46d","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-08 08:25:41.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8fcfff20fe0b5b37d7c55ee0abcc7e824b283f58","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-08-08 20:14:30.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"de8da0f0363ad680d1bcbb8ea3ac9eb1bb033420","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-09 08:13:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"780807b39b73ec28ae13d469671dc2165e11375b","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-09 08:15:13.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"46f4a34e3c7b9ab33b60c432d8296b75f5368c8d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-08-09 08:22:05.000000000","message":"Patch Set 11: Code-Review+1\n\n(2 comments)\n\nPlease check my comments in PS10","accounts_in_message":[],"_revision_number":11},{"id":"c59a6e9de890e424432b15b854460eb1b3fbbad2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-09 10:07:16.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/20/577220/11/check/requirements-check/5b6275d/ : SUCCESS in 4m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/11/check/build-openstack-sphinx-docs/7ab16f3/html/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/11/check/openstack-tox-pep8/7fbe4f6/ : SUCCESS in 5m 33s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/11/check/openstack-tox-py27/379247f/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/11/check/openstack-tox-py35/2ba7ef1/ : SUCCESS in 5m 51s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/11/check/legacy-tempest-dsvm-neutron-src/94ce108/ : SUCCESS in 1h 41m 23s\n- openstack-tox-cover http://logs.openstack.org/20/577220/11/check/openstack-tox-cover/02b749e/cover/ : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/11/check/openstack-tox-lower-constraints/a72da9a/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":11},{"id":"81299d397d255987b8008ed327a853fcb0b63c5b","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2018-08-09 18:19:48.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"aa15558f484c3a772386073dcb0f9a7392a6a92b","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-10 08:53:39.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9250fde07e18460f70f330573e38a3534d1186c8","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-10 08:55:35.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d574a2e32798838a20d94172cd5677b2cbb0193a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-10 10:15:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/20/577220/12/check/requirements-check/b013ef6/ : SUCCESS in 4m 37s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/12/check/build-openstack-sphinx-docs/e94e149/html/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/12/check/openstack-tox-pep8/7ad3534/ : SUCCESS in 7m 33s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/12/check/openstack-tox-py27/d2feb55/ : SUCCESS in 7m 15s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/12/check/openstack-tox-py35/f60b080/ : SUCCESS in 5m 05s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/12/check/legacy-tempest-dsvm-neutron-src/df7d522/ : SUCCESS in 1h 20m 29s\n- openstack-tox-cover http://logs.openstack.org/20/577220/12/check/openstack-tox-cover/e914db8/cover/ : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/12/check/openstack-tox-lower-constraints/ab34ebc/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":12},{"id":"396383a6873d9a5009dab9389bb7ec10dce8447a","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-08-10 19:01:49.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"139d92c82266147b087a4c00653b59b89555f5a9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-08-13 07:34:15.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7ab2b18b359e6ea314fb0975b651c437416e48c8","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2018-08-13 15:11:06.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"0178cf248f042ee4b71dbaa17d216f6607a53580","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2018-08-14 13:42:03.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"11246719c9dd7d82169e0787381b18e6812b293e","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-15 09:42:51.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"510c79bad15b76f7e916d1b4aa9bfebb1a296bd6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-15 11:10:44.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/20/577220/13/check/requirements-check/dcc0197/ : SUCCESS in 4m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/13/check/build-openstack-sphinx-docs/f9346e8/html/ : SUCCESS in 4m 51s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/13/check/openstack-tox-pep8/8564562/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/13/check/openstack-tox-py27/b81d4b9/ : SUCCESS in 5m 10s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/13/check/openstack-tox-py35/5a35b94/ : SUCCESS in 5m 47s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/13/check/legacy-tempest-dsvm-neutron-src/6ee1060/ : SUCCESS in 1h 25m 31s\n- openstack-tox-cover http://logs.openstack.org/20/577220/13/check/openstack-tox-cover/be326d2/cover/ : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/13/check/openstack-tox-lower-constraints/a8a33de/ : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":13},{"id":"424620e3ef16a22326ed69fe4385f6434e839bc1","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2018-08-15 14:28:14.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)\n\nPlease include a release note","accounts_in_message":[],"_revision_number":13},{"id":"9260d503d78afc781c7f63ac3f7de1c0d7f83650","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-08-15 19:43:53.000000000","message":"Removed Code-Review+2 by Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":13},{"id":"fb4547f97a210350073afd889ed25f90cfd80d9f","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-16 13:47:55.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"7223dd770c4a265391258593abbbd0c96f5946b6","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-16 13:52:12.000000000","message":"Patch Set 14:\n\n(2 comments)\n\nAdded a release note.","accounts_in_message":[],"_revision_number":14},{"id":"4e1afa752a7dc9292a2ccbde77e8e70458211833","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-16 15:18:08.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/20/577220/14/check/requirements-check/2190ac9/ : SUCCESS in 4m 18s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/14/check/build-openstack-sphinx-docs/353df2f/html/ : SUCCESS in 6m 14s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/14/check/openstack-tox-pep8/1bdb28b/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/14/check/openstack-tox-py27/b8141c3/ : SUCCESS in 5m 23s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/14/check/openstack-tox-py35/d3d94cf/ : SUCCESS in 5m 51s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/14/check/legacy-tempest-dsvm-neutron-src/c4d83e5/ : SUCCESS in 1h 26m 50s\n- build-openstack-releasenotes http://logs.openstack.org/20/577220/14/check/build-openstack-releasenotes/da702db/html/ : SUCCESS in 5m 29s\n- openstack-tox-cover http://logs.openstack.org/20/577220/14/check/openstack-tox-cover/12f5700/cover/ : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/14/check/openstack-tox-lower-constraints/bc7b6b4/ : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":14},{"id":"c931f34a405c37fe127528e473950300cb6b0809","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-08-16 18:47:40.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"b04b8a42a034c2f80b5c3f3f5f68c24875da5da7","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-17 07:18:21.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"cb29ce971944405617a1c75981638f591959f7f1","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-17 07:19:17.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"ea20ef3398ca60173957798358b094c8b1644764","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-17 08:43:11.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/20/577220/15/check/requirements-check/0618133/ : SUCCESS in 3m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/15/check/build-openstack-sphinx-docs/e5356d4/html/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/15/check/openstack-tox-pep8/a9ed333/ : FAILURE in 5m 06s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/15/check/openstack-tox-py27/5cc7f11/ : FAILURE in 5m 36s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/15/check/openstack-tox-py35/54609be/ : FAILURE in 5m 20s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/15/check/legacy-tempest-dsvm-neutron-src/15ad519/ : SUCCESS in 1h 20m 23s\n- build-openstack-releasenotes http://logs.openstack.org/20/577220/15/check/build-openstack-releasenotes/49b67e1/html/ : SUCCESS in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/20/577220/15/check/openstack-tox-cover/c56aca1/ : FAILURE in 5m 01s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/15/check/openstack-tox-lower-constraints/460715a/ : FAILURE in 5m 24s","accounts_in_message":[],"_revision_number":15},{"id":"27847d06cbe861422432091759864a2a51a2f10c","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-18 05:19:56.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"c3beb026aa7becc88b8f548f23859d28acc1b59d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-18 06:47:04.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/20/577220/16/check/requirements-check/5274488/ : SUCCESS in 4m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/16/check/build-openstack-sphinx-docs/becbd9a/html/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/16/check/openstack-tox-pep8/a251101/ : FAILURE in 4m 54s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/16/check/openstack-tox-py27/e1dc8e3/ : FAILURE in 4m 36s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/16/check/openstack-tox-py35/e5a9fc2/ : FAILURE in 4m 53s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/16/check/legacy-tempest-dsvm-neutron-src/d23c1c5/ : SUCCESS in 1h 25m 15s\n- build-openstack-releasenotes http://logs.openstack.org/20/577220/16/check/build-openstack-releasenotes/6d34382/html/ : SUCCESS in 3m 53s\n- openstack-tox-cover http://logs.openstack.org/20/577220/16/check/openstack-tox-cover/354b4ba/ : FAILURE in 4m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/16/check/openstack-tox-lower-constraints/65e7e01/ : FAILURE in 4m 59s","accounts_in_message":[],"_revision_number":16},{"id":"6b4ed6e037dfa58f29882fa5dd5d37e1af5327de","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-08-20 20:05:25.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"64cd22f29c60d0d356f00cf4cda3a4eb9f117991","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-21 14:55:08.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"7dab6975dd272eb738040b83bf56c8463d63532a","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-21 14:59:45.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"afb2c0f7437bf825d5e52fe015ea45e8484b7b8f","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-08-21 17:08:09.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"5a7f3e01a1d58efdd0fb7b84953dd65c547f25f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-21 17:19:13.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/20/577220/17/check/requirements-check/a48954d/ : SUCCESS in 3m 29s\n- build-openstack-sphinx-docs finger://ze01.openstack.org/e881d0caf19d453f9b6c2e1bce320ccf : RETRY_LIMIT in 1m 00s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/17/check/openstack-tox-pep8/27ef35c/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/17/check/openstack-tox-py27/38f05da/ : SUCCESS in 5m 55s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/17/check/openstack-tox-py35/5623f11/ : SUCCESS in 5m 22s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/17/check/legacy-tempest-dsvm-neutron-src/1a219a8/ : SUCCESS in 1h 59m 31s\n- build-openstack-releasenotes http://logs.openstack.org/20/577220/17/check/build-openstack-releasenotes/5037bfb/html/ : SUCCESS in 4m 41s\n- openstack-tox-cover http://logs.openstack.org/20/577220/17/check/openstack-tox-cover/508bdbd/cover/ : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/17/check/openstack-tox-lower-constraints/3053677/ : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":17},{"id":"02fa2d3164b44492be1058fbef9cb3afa38d1a48","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2018-08-22 07:58:18.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"a1885a39c7d3d98a1822f482540e1972cb26c6c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-22 09:26:11.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/20/577220/17/check/requirements-check/ac484da/ : SUCCESS in 3m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/17/check/build-openstack-sphinx-docs/abb52a8/html/ : SUCCESS in 4m 31s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/17/check/openstack-tox-pep8/52cfa57/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/17/check/openstack-tox-py27/75e500a/ : SUCCESS in 5m 43s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/17/check/openstack-tox-py35/b995def/ : SUCCESS in 5m 11s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/17/check/legacy-tempest-dsvm-neutron-src/f4ab8c6/ : SUCCESS in 1h 25m 33s\n- build-openstack-releasenotes http://logs.openstack.org/20/577220/17/check/build-openstack-releasenotes/c1ae466/html/ : SUCCESS in 4m 03s\n- openstack-tox-cover http://logs.openstack.org/20/577220/17/check/openstack-tox-cover/d4546b8/cover/ : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/17/check/openstack-tox-lower-constraints/470eef7/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":17},{"id":"b4b77e3e17863686f804f71660cdc1e5318dd4b1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2018-08-22 09:32:14.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"295c1a717e6f04e722e11283433f171d4a079794","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2018-08-22 13:52:13.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"10f9f65909410b708aa796b2ab82dfecc521db68","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2018-08-22 14:24:19.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"74b22a53af4b3c6e2057d0007ccf0fea985ded9b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-08-22 19:40:59.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"689d21b504ee647da8afdf8cc180ac4313cedf07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-22 19:41:11.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"85b25b4dfa976edb3e57f83120573d3a697b0efa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-22 21:05:12.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/20/577220/17/gate/requirements-check/18807ab/ : SUCCESS in 4m 01s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/577220/17/gate/build-openstack-sphinx-docs/986b989/html/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/20/577220/17/gate/openstack-tox-pep8/c8e0bb5/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/20/577220/17/gate/openstack-tox-py27/76e50b1/ : SUCCESS in 5m 20s\n- openstack-tox-py35 http://logs.openstack.org/20/577220/17/gate/openstack-tox-py35/4eb84dd/ : SUCCESS in 5m 17s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/20/577220/17/gate/legacy-tempest-dsvm-neutron-src/f283b00/ : SUCCESS in 1h 21m 23s\n- build-openstack-releasenotes http://logs.openstack.org/20/577220/17/gate/build-openstack-releasenotes/749c0c0/html/ : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/20/577220/17/gate/openstack-tox-lower-constraints/3f26aba/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":17},{"id":"a07292a348926172ed3952edb5dd439b7d5ac6cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-22 21:05:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"579e0ccabbd5ec132366fe51f46be653cbf15986","revisions":{"deb00875cbcefc0321053ed040f2d4d82fcc3b88":{"kind":"REWORK","_number":1,"created":"2018-06-21 15:22:37.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/1"}}},"commit":{"parents":[{"commit":"f5978ab6d73be40164e4bf915939f4f8d60281e4","subject":"Merge \"rehome port-security-groups-filtering api extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f5978ab6d73be40164e4bf915939f4f8d60281e4"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-06-18 15:19:36.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-06-21 15:20:47.000000000","tz":120},"subject":"WIP Placement: constants and utils","message":"WIP Placement: constants and utils\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: Ib197573e5cdb60ef0db4e7a771c3179bf9d5bb95\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/deb00875cbcefc0321053ed040f2d4d82fcc3b88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/deb00875cbcefc0321053ed040f2d4d82fcc3b88"}]},"branch":"refs/heads/master"},"5d1cf42871e802563f809e7ca43bfe5f69275bc5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-06-22 11:40:40.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/2"}}},"commit":{"parents":[{"commit":"f5978ab6d73be40164e4bf915939f4f8d60281e4","subject":"Merge \"rehome port-security-groups-filtering api extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f5978ab6d73be40164e4bf915939f4f8d60281e4"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-06-18 15:19:36.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-06-22 07:30:24.000000000","tz":120},"subject":"WIP Placement: constants and utils","message":"WIP Placement: constants and utils\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5d1cf42871e802563f809e7ca43bfe5f69275bc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5d1cf42871e802563f809e7ca43bfe5f69275bc5"}]},"branch":"refs/heads/master"},"e0fccc8145ae5e99cc0ff7bd738947df520b78b2":{"kind":"REWORK","_number":3,"created":"2018-07-06 14:05:55.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/3"}}},"commit":{"parents":[{"commit":"d32092e06aef67cb56c5882135a6c8b32b6ef51d","subject":"WIP Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d32092e06aef67cb56c5882135a6c8b32b6ef51d"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 14:04:14.000000000","tz":120},"subject":"WIP Placement: utils","message":"WIP Placement: utils\n\nTODO tests\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e0fccc8145ae5e99cc0ff7bd738947df520b78b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e0fccc8145ae5e99cc0ff7bd738947df520b78b2"}]},"branch":"refs/heads/master"},"f9505e9e76f48175caa72c121868ae57822068c8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-07-09 09:29:58.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/4"}}},"commit":{"parents":[{"commit":"a953fdca9651d9e9a411852a0c7586a6db95e826","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a953fdca9651d9e9a411852a0c7586a6db95e826"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-09 09:29:06.000000000","tz":120},"subject":"WIP Placement: utils","message":"WIP Placement: utils\n\nTODO tests\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f9505e9e76f48175caa72c121868ae57822068c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f9505e9e76f48175caa72c121868ae57822068c8"}]},"branch":"refs/heads/master"},"4fd5cf56172bed34a3059f4beaf70375a780aa5c":{"kind":"REWORK","_number":5,"created":"2018-07-10 12:48:45.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/5"}}},"commit":{"parents":[{"commit":"a953fdca9651d9e9a411852a0c7586a6db95e826","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a953fdca9651d9e9a411852a0c7586a6db95e826"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-10 12:40:40.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4fd5cf56172bed34a3059f4beaf70375a780aa5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4fd5cf56172bed34a3059f4beaf70375a780aa5c"}]},"branch":"refs/heads/master"},"06500cb3dd39b12c02203131b5d48feac0f36ced":{"kind":"REWORK","_number":6,"created":"2018-07-10 12:49:57.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/6"}}},"commit":{"parents":[{"commit":"a953fdca9651d9e9a411852a0c7586a6db95e826","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a953fdca9651d9e9a411852a0c7586a6db95e826"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-10 12:49:32.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/06500cb3dd39b12c02203131b5d48feac0f36ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/06500cb3dd39b12c02203131b5d48feac0f36ced"}]},"branch":"refs/heads/master"},"fa1db2452473a930faf1e0d1b1452822197a904b":{"kind":"REWORK","_number":7,"created":"2018-07-12 13:55:01.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/7"}}},"commit":{"parents":[{"commit":"a953fdca9651d9e9a411852a0c7586a6db95e826","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a953fdca9651d9e9a411852a0c7586a6db95e826"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-12 13:53:51.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fa1db2452473a930faf1e0d1b1452822197a904b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fa1db2452473a930faf1e0d1b1452822197a904b"}]},"branch":"refs/heads/master"},"3b55e6db9be5b927c4b5098f8f559e7e0062124b":{"kind":"REWORK","_number":8,"created":"2018-07-20 11:53:06.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/8"}}},"commit":{"parents":[{"commit":"353ac606d8974914918e80ad0f38840ad0cc03e3","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/353ac606d8974914918e80ad0f38840ad0cc03e3"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-20 11:46:15.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b55e6db9be5b927c4b5098f8f559e7e0062124b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b55e6db9be5b927c4b5098f8f559e7e0062124b"}]},"branch":"refs/heads/master"},"97a46f3769b4a1953f694520eed225621215e220":{"kind":"REWORK","_number":9,"created":"2018-08-01 10:53:10.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/9"}}},"commit":{"parents":[{"commit":"238e4f7c476a0e60f73441df3d3ea3ed79457070","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/238e4f7c476a0e60f73441df3d3ea3ed79457070"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-08-01 10:50:58.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/97a46f3769b4a1953f694520eed225621215e220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/97a46f3769b4a1953f694520eed225621215e220"}]},"branch":"refs/heads/master"},"1ad030b887b8765339eba4a43d9c180c435a7f6e":{"kind":"REWORK","_number":10,"created":"2018-08-06 13:01:58.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/10"}}},"commit":{"parents":[{"commit":"238e4f7c476a0e60f73441df3d3ea3ed79457070","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/238e4f7c476a0e60f73441df3d3ea3ed79457070"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-08-06 13:01:09.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1ad030b887b8765339eba4a43d9c180c435a7f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1ad030b887b8765339eba4a43d9c180c435a7f6e"}]},"branch":"refs/heads/master"},"4f5765c6616f4c1b4b6c73a5782334ed2e939955":{"kind":"REWORK","_number":11,"created":"2018-08-09 08:13:19.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/11"}}},"commit":{"parents":[{"commit":"3b58c700639ce294f9b102de84d99e69289246c3","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b58c700639ce294f9b102de84d99e69289246c3"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-08-09 08:08:45.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4f5765c6616f4c1b4b6c73a5782334ed2e939955"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4f5765c6616f4c1b4b6c73a5782334ed2e939955"}]},"branch":"refs/heads/master"},"6811e4256a5af5175090759248224751c24cfda8":{"kind":"REWORK","_number":12,"created":"2018-08-10 08:53:39.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/12"}}},"commit":{"parents":[{"commit":"3b58c700639ce294f9b102de84d99e69289246c3","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b58c700639ce294f9b102de84d99e69289246c3"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-08-10 08:53:02.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6811e4256a5af5175090759248224751c24cfda8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6811e4256a5af5175090759248224751c24cfda8"}]},"branch":"refs/heads/master"},"51687f5188653416ad3f810c83d929496fead077":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2018-08-15 09:42:51.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/13"}}},"commit":{"parents":[{"commit":"8b6b841269c639f76b3f23857ea1352b77eee772","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8b6b841269c639f76b3f23857ea1352b77eee772"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-08-15 09:41:01.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/51687f5188653416ad3f810c83d929496fead077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/51687f5188653416ad3f810c83d929496fead077"}]},"branch":"refs/heads/master"},"f4b0dbfd00d4f6144a42e3080f51ec209f7de095":{"kind":"REWORK","_number":14,"created":"2018-08-16 13:47:55.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/14"}}},"commit":{"parents":[{"commit":"7e340fd3275620a8ab10777e3d8bfec68807f572","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e340fd3275620a8ab10777e3d8bfec68807f572"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-08-16 13:39:07.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f4b0dbfd00d4f6144a42e3080f51ec209f7de095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f4b0dbfd00d4f6144a42e3080f51ec209f7de095"}]},"branch":"refs/heads/master"},"97816bddfc9a26213999f3359e38a687dc0eaf4a":{"kind":"REWORK","_number":15,"created":"2018-08-17 07:18:21.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/15"}}},"commit":{"parents":[{"commit":"7e340fd3275620a8ab10777e3d8bfec68807f572","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e340fd3275620a8ab10777e3d8bfec68807f572"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-08-17 07:15:37.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/97816bddfc9a26213999f3359e38a687dc0eaf4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/97816bddfc9a26213999f3359e38a687dc0eaf4a"}]},"branch":"refs/heads/master"},"8ab0aebfee9d18ea40bad40efdcd9dcbe0d16c06":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2018-08-18 05:19:56.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/16"}}},"commit":{"parents":[{"commit":"b76931f29b468b6427950ae5d31ee5fa8f0503dd","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b76931f29b468b6427950ae5d31ee5fa8f0503dd"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-08-18 05:19:56.000000000","tz":0},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ab0aebfee9d18ea40bad40efdcd9dcbe0d16c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8ab0aebfee9d18ea40bad40efdcd9dcbe0d16c06"}]},"branch":"refs/heads/master"},"579e0ccabbd5ec132366fe51f46be653cbf15986":{"kind":"REWORK","_number":17,"created":"2018-08-21 14:55:08.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/20/577220/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/577220/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/577220/17"}}},"commit":{"parents":[{"commit":"b271a336b13c0d8d6b5ef0078412d87040094747","subject":"Placement: constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b271a336b13c0d8d6b5ef0078412d87040094747"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-07-06 13:33:12.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@ericsson.com","date":"2018-08-21 14:52:30.000000000","tz":120},"subject":"Placement: utils","message":"Placement: utils\n\n* to generate Placement trait names,\n* to generate Placement resource provider UUIDs, and\n* to uniformly parse and validate Placement-related config options in\n  all agents.\n\nChange-Id: I192d99673feba97a95af995923b266e2f8b58c6d\nNeeded-By: https://review.openstack.org/577223\nPartial-Bug: #1578989\nSee-Also: https://review.openstack.org/502306 (nova spec)\nSee-Also: https://review.openstack.org/508149 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/579e0ccabbd5ec132366fe51f46be653cbf15986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/579e0ccabbd5ec132366fe51f46be653cbf15986"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
