)]}'
{"id":"openstack%2Fneutron-lib~577866","triplet_id":"openstack%2Fneutron-lib~master~I64780fc435da98d9b5d4e58c9e61d3d0b11d0daf","project":"openstack/neutron-lib","branch":"master","hashtags":[],"change_id":"I64780fc435da98d9b5d4e58c9e61d3d0b11d0daf","subject":"Update the segment api extension","status":"MERGED","created":"2018-06-25 16:41:10.000000000","updated":"2018-08-22 18:24:24.000000000","submitted":"2018-08-22 18:24:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"577866-1534962264486-865d8286","meta_rev_id":"ea9113d5f5656bb97cfca78d32f0c9c5d958fabf","_number":577866,"virtual_id_number":577866,"owner":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":7715,"name":"Hirofumi Ichihara","email":"ichihara.hirofumi@gmail.com","username":"Hirofumi-Ichihara"},{"value":0,"date":"2018-08-13 15:50:11.000000000","_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2018-08-22 18:24:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":7715,"name":"Hirofumi Ichihara","email":"ichihara.hirofumi@gmail.com","username":"Hirofumi-Ichihara"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":2,"date":"2018-08-22 16:59:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2018-08-22 14:42:11.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":7715,"name":"Hirofumi Ichihara","email":"ichihara.hirofumi@gmail.com","username":"Hirofumi-Ichihara"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":1,"date":"2018-08-22 16:59:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":7715,"name":"Hirofumi Ichihara","email":"ichihara.hirofumi@gmail.com","username":"Hirofumi-Ichihara"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":7715,"name":"Hirofumi Ichihara","email":"ichihara.hirofumi@gmail.com","username":"Hirofumi-Ichihara"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-25 19:57:49.000000000","updated_by":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"reviewer":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"state":"REVIEWER"},{"updated":"2018-07-13 15:06:18.000000000","updated_by":{"_account_id":7715,"name":"Hirofumi Ichihara","email":"ichihara.hirofumi@gmail.com","username":"Hirofumi-Ichihara"},"reviewer":{"_account_id":7715,"name":"Hirofumi Ichihara","email":"ichihara.hirofumi@gmail.com","username":"Hirofumi-Ichihara"},"state":"REVIEWER"},{"updated":"2018-08-13 08:14:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2018-08-22 14:42:11.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2018-08-22 16:59:26.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2018-08-22 18:24:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fffa329eff193965139a1aa3658bae5cca8c2b5f","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-06-25 16:41:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38dd9917b1e0734dade4275a996701990f405e3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-25 18:20:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/66/577866/1/check/build-openstack-sphinx-docs/c156776/html/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 http://logs.openstack.org/66/577866/1/check/openstack-tox-pep8/bc8ba0e/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/66/577866/1/check/openstack-tox-py27/e75f5bb/ : SUCCESS in 5m 56s\n- openstack-tox-py35 http://logs.openstack.org/66/577866/1/check/openstack-tox-py35/00b9677/ : SUCCESS in 5m 13s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/66/577866/1/check/legacy-tempest-dsvm-neutron-src/6cffcdf/ : SUCCESS in 1h 23m 55s\n- openstack-tox-cover http://logs.openstack.org/66/577866/1/check/openstack-tox-cover/6b51730/cover/ : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/66/577866/1/check/openstack-tox-lower-constraints/008fde5/ : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b6adc9f0173695b5e9bf84d9d796541c11a02ece","author":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"date":"2018-06-25 19:57:49.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"20e1a2cf27edf4358f23df4b715701cc5e971c85","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2018-06-26 10:50:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f29f7c0308c3f3e89c21f616ceb93faad37f3771","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2018-06-26 10:50:46.000000000","message":"Patch Set 1:\n\nAlso borderline as to if a release note is required... Might be nice to include one.","accounts_in_message":[],"_revision_number":1},{"id":"7d21e9b93b8436892cf90aeb4eded15a809ae617","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-06-26 22:15:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cee6a442e2f80c31222f8bb94033101f9c9691ce","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-06-26 22:16:20.000000000","message":"Patch Set 2:\n\n(3 comments)\n\n\u003e Also borderline as to if a release note is required... Might be\n \u003e nice to include one.\n\nDone","accounts_in_message":[],"_revision_number":2},{"id":"700a70b0b54fb0f13e6afe764548f150fdfb9e90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-27 01:04:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/66/577866/2/check/build-openstack-sphinx-docs/b12f5c9/html/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/66/577866/2/check/openstack-tox-pep8/4ddab35/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/66/577866/2/check/openstack-tox-py27/b401a47/ : SUCCESS in 4m 52s\n- openstack-tox-py35 http://logs.openstack.org/66/577866/2/check/openstack-tox-py35/7ce503c/ : SUCCESS in 5m 02s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/66/577866/2/check/legacy-tempest-dsvm-neutron-src/f92aad4/ : SUCCESS in 1h 35m 46s\n- build-openstack-releasenotes http://logs.openstack.org/66/577866/2/check/build-openstack-releasenotes/2d7398f/html/ : SUCCESS in 4m 07s\n- openstack-tox-cover http://logs.openstack.org/66/577866/2/check/openstack-tox-cover/54ba724/cover/ : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/66/577866/2/check/openstack-tox-lower-constraints/e062960/ : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"8626a50db87c3bc2342af774d3a6b40172226321","author":{"_account_id":7715,"name":"Hirofumi Ichihara","email":"ichihara.hirofumi@gmail.com","username":"Hirofumi-Ichihara"},"date":"2018-06-29 02:37:46.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7146a2671c3a0427cc6169d90f58f6005aefccbc","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-06-29 14:35:25.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6d774471edefd42eef7c62f93fa9f353be1fa808","author":{"_account_id":7715,"name":"Hirofumi Ichihara","email":"ichihara.hirofumi@gmail.com","username":"Hirofumi-Ichihara"},"date":"2018-07-02 02:58:58.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b7916e4beee145ba0fdb25d2f63590eb2524d47","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2018-07-02 13:11:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"80ad7aaa9f37dcab31ca4fcf9c5e465e700c2de3","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-07-03 19:56:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7a29c16997bde22861ac4eb31b14b5dfbed957b","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2018-07-06 13:13:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb5f9aef7ccc3e403997f37e3be237b0b492b602","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-07-06 14:57:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f08ef5b997e34df3e7a61a3b41c950387a30f482","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-07-06 14:58:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d8de1e1f4e496a6d8c7e87d4febd9bdd598ae0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-06 15:22:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/66/577866/3/check/build-openstack-sphinx-docs/5ff1b56/html/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/66/577866/3/check/openstack-tox-pep8/853f838/ : SUCCESS in 6m 23s\n- openstack-tox-py27 http://logs.openstack.org/66/577866/3/check/openstack-tox-py27/f33cef9/ : FAILURE in 5m 28s\n- openstack-tox-py35 http://logs.openstack.org/66/577866/3/check/openstack-tox-py35/bd9272f/ : FAILURE in 5m 07s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/66/577866/3/check/legacy-tempest-dsvm-neutron-src/87f96e2/ : FAILURE in 21m 35s\n- build-openstack-releasenotes http://logs.openstack.org/66/577866/3/check/build-openstack-releasenotes/4e88530/html/ : SUCCESS in 3m 37s\n- openstack-tox-cover http://logs.openstack.org/66/577866/3/check/openstack-tox-cover/8c03df5/ : FAILURE in 5m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/66/577866/3/check/openstack-tox-lower-constraints/2c4ea62/ : FAILURE in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"c94ffdd7d3f48455c14a8cbdae2ba7c61f7f58fd","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-07-06 15:55:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bd54c1db12ad7c83444a6407c98ab25dbd9c4fec","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-07-06 15:56:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6bf8e557e8051628e18ea5f4d949bcf0c7d3740b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-06 17:23:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/66/577866/4/check/build-openstack-sphinx-docs/05a6018/html/ : SUCCESS in 5m 08s\n- openstack-tox-pep8 http://logs.openstack.org/66/577866/4/check/openstack-tox-pep8/187f580/ : SUCCESS in 6m 30s\n- openstack-tox-py27 http://logs.openstack.org/66/577866/4/check/openstack-tox-py27/b2675e1/ : SUCCESS in 8m 00s\n- openstack-tox-py35 http://logs.openstack.org/66/577866/4/check/openstack-tox-py35/da03ecd/ : SUCCESS in 6m 13s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/66/577866/4/check/legacy-tempest-dsvm-neutron-src/bc50b3e/ : SUCCESS in 1h 16m 56s\n- build-openstack-releasenotes http://logs.openstack.org/66/577866/4/check/build-openstack-releasenotes/9344fc0/html/ : SUCCESS in 6m 33s\n- openstack-tox-cover http://logs.openstack.org/66/577866/4/check/openstack-tox-cover/8b4c130/cover/ : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/66/577866/4/check/openstack-tox-lower-constraints/a841138/ : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":4},{"id":"b7e713a715c35ff987632dd304c602f227565094","author":{"_account_id":7715,"name":"Hirofumi Ichihara","email":"ichihara.hirofumi@gmail.com","username":"Hirofumi-Ichihara"},"date":"2018-07-13 15:06:18.000000000","message":"Patch Set 4:\n\nIMO, segment extension didn\u0027t have to require standard-attr-segment[1] because of backward compatibility. It means that standard attributes are added to original segment API even if existing segment users don\u0027t want it. My idea is removing [1] from neutron side.\n\n[1]: https://github.com/openstack/neutron/blob/master/neutron/extensions/segment.py#L136","accounts_in_message":[],"_revision_number":4},{"id":"35a63226abb143850d138d0588583423bdce80b1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2018-08-13 08:14:50.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a51d6be6b974ccee91998e17d923c6189a5d1d64","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-08-13 15:47:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f7a1334ac6210a1fe19b1a555fe00494e5a8871e","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2018-08-13 15:50:11.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4729d790d0190bb2f7f45ca8ab6b03ca7d61f407","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-13 17:13:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/66/577866/5/check/build-openstack-sphinx-docs/ad13b4d/html/ : SUCCESS in 5m 29s\n- openstack-tox-pep8 http://logs.openstack.org/66/577866/5/check/openstack-tox-pep8/7006dc6/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/66/577866/5/check/openstack-tox-py27/7acad13/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/66/577866/5/check/openstack-tox-py35/791228f/ : SUCCESS in 6m 12s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/66/577866/5/check/legacy-tempest-dsvm-neutron-src/57c52b2/ : SUCCESS in 1h 22m 48s\n- build-openstack-releasenotes http://logs.openstack.org/66/577866/5/check/build-openstack-releasenotes/85053d7/html/ : SUCCESS in 4m 11s\n- openstack-tox-cover http://logs.openstack.org/66/577866/5/check/openstack-tox-cover/efecdcb/cover/ : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/66/577866/5/check/openstack-tox-lower-constraints/0d22ca7/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":5},{"id":"984debfdf21e8374e99de9e80c27f4ee89cc3dd0","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2018-08-22 14:42:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"de0c219fa60096e1d62091a66b491f3c790d26a2","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2018-08-22 16:59:26.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"98eb95750f1b298ab292404d76cf246cd4e085f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-22 16:59:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f837aafc8fe25d7339124ad530909a6afb9c8fe4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-22 18:24:24.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/66/577866/5/gate/build-openstack-sphinx-docs/051f84e/html/ : SUCCESS in 5m 19s\n- openstack-tox-pep8 http://logs.openstack.org/66/577866/5/gate/openstack-tox-pep8/17d47e6/ : SUCCESS in 5m 51s\n- openstack-tox-py27 http://logs.openstack.org/66/577866/5/gate/openstack-tox-py27/6556cf9/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/66/577866/5/gate/openstack-tox-py35/3039d17/ : SUCCESS in 5m 13s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/66/577866/5/gate/legacy-tempest-dsvm-neutron-src/3fdf4b1/ : SUCCESS in 1h 21m 48s\n- build-openstack-releasenotes http://logs.openstack.org/66/577866/5/gate/build-openstack-releasenotes/4d9de5c/html/ : SUCCESS in 3m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/577866/5/gate/openstack-tox-lower-constraints/15294e7/ : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":5},{"id":"ea9113d5f5656bb97cfca78d32f0c9c5d958fabf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-22 18:24:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e3d90fef655b41c36f0e6401a3b19118d496d99c","revisions":{"88d53455baed35c42fd2fe379a48fb14b7c3042a":{"kind":"REWORK","_number":1,"created":"2018-06-25 16:41:10.000000000","uploader":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"ref":"refs/changes/66/577866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/577866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/1"}}},"commit":{"parents":[{"commit":"8664953c5a17e8abf5c5a52832e63885108aa818","subject":"Add release notes link to README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8664953c5a17e8abf5c5a52832e63885108aa818"}]}],"author":{"name":"Hongbin Lu","email":"hongbin.lu@huawei.com","date":"2018-06-25 16:37:52.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin.lu@huawei.com","date":"2018-06-25 16:37:52.000000000","tz":0},"subject":"Update the segment api extension","message":"Update the segment api extension\n\nThe segment api defintion is changed in server side [1]. This patch\nsync up the defintion in neutron-lib. In addition, remove the\ndependency from standard-attr-segment on segment since the\ndependency is reserved: we have segment depending on\nstandard-attr-segment with the latest change.\n\n[1] https://review.openstack.org/#/c/571292/\n\nChange-Id: I64780fc435da98d9b5d4e58c9e61d3d0b11d0daf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/88d53455baed35c42fd2fe379a48fb14b7c3042a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/88d53455baed35c42fd2fe379a48fb14b7c3042a"}]},"branch":"refs/heads/master"},"f774221409e10ce186b4535b3199a419b82501a8":{"kind":"REWORK","_number":2,"created":"2018-06-26 22:15:44.000000000","uploader":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"ref":"refs/changes/66/577866/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/577866/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/2"}}},"commit":{"parents":[{"commit":"76020dd81f0a65bca34abc9d02de0e48b09545d4","subject":"Merge \"Add a shim extension availability_zone_filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/76020dd81f0a65bca34abc9d02de0e48b09545d4"}]}],"author":{"name":"Hongbin Lu","email":"hongbin.lu@huawei.com","date":"2018-06-25 16:37:52.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin.lu@huawei.com","date":"2018-06-26 22:15:37.000000000","tz":0},"subject":"Update the segment api extension","message":"Update the segment api extension\n\nThe segment api definition is changed in server side [1]. This patch\nsync up the definition in neutron-lib. In addition, remove the\ndependency from standard-attr-segment on segment since the\ndependency is reserved: we have segment depending on\nstandard-attr-segment with the latest change.\n\n[1] https://review.openstack.org/#/c/571292/\n\nChange-Id: I64780fc435da98d9b5d4e58c9e61d3d0b11d0daf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f774221409e10ce186b4535b3199a419b82501a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f774221409e10ce186b4535b3199a419b82501a8"}]},"branch":"refs/heads/master"},"03d06f5cf14887fb686a190e32b583f8b9eeece3":{"kind":"REWORK","_number":3,"created":"2018-07-06 14:57:07.000000000","uploader":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"ref":"refs/changes/66/577866/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/577866/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/3"}}},"commit":{"parents":[{"commit":"8cbab5756c7b6adffc54af39ead3b6c89bbcd3ea","subject":"Merge \"Cleanup unused key-value in the attribute of l3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8cbab5756c7b6adffc54af39ead3b6c89bbcd3ea"}]}],"author":{"name":"Hongbin Lu","email":"hongbin.lu@huawei.com","date":"2018-06-25 16:37:52.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin.lu@huawei.com","date":"2018-07-06 14:57:02.000000000","tz":0},"subject":"Update the segment api extension","message":"Update the segment api extension\n\nThe segment api definition is changed in server side [1]. This patch\nsync up the definition in neutron-lib. In addition, remove the\ndependency from standard-attr-segment on segment since the\ndependency is reserved: we have segment depending on\nstandard-attr-segment with the latest change.\n\n[1] https://review.openstack.org/#/c/571292/\n\nChange-Id: I64780fc435da98d9b5d4e58c9e61d3d0b11d0daf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/03d06f5cf14887fb686a190e32b583f8b9eeece3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/03d06f5cf14887fb686a190e32b583f8b9eeece3"}]},"branch":"refs/heads/master"},"fff9a49691816f902edd6cc149c30d56229c76a4":{"kind":"REWORK","_number":4,"created":"2018-07-06 15:55:54.000000000","uploader":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"ref":"refs/changes/66/577866/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/577866/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/4"}}},"commit":{"parents":[{"commit":"8cbab5756c7b6adffc54af39ead3b6c89bbcd3ea","subject":"Merge \"Cleanup unused key-value in the attribute of l3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8cbab5756c7b6adffc54af39ead3b6c89bbcd3ea"}]}],"author":{"name":"Hongbin Lu","email":"hongbin.lu@huawei.com","date":"2018-06-25 16:37:52.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin.lu@huawei.com","date":"2018-07-06 15:55:49.000000000","tz":0},"subject":"Update the segment api extension","message":"Update the segment api extension\n\nThe segment api definition is changed in server side [1]. This patch\nsync up the definition in neutron-lib. In addition, remove the\ndependency from standard-attr-segment on segment since the\ndependency is reserved: we have segment depending on\nstandard-attr-segment with the latest change.\n\n[1] https://review.openstack.org/#/c/571292/\n\nChange-Id: I64780fc435da98d9b5d4e58c9e61d3d0b11d0daf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fff9a49691816f902edd6cc149c30d56229c76a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fff9a49691816f902edd6cc149c30d56229c76a4"}]},"branch":"refs/heads/master"},"e3d90fef655b41c36f0e6401a3b19118d496d99c":{"kind":"REWORK","_number":5,"created":"2018-08-13 15:47:27.000000000","uploader":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"ref":"refs/changes/66/577866/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/66/577866/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/66/577866/5"}}},"commit":{"parents":[{"commit":"34af1e37bb417b61110695d88db6186a8aece02f","subject":"api-ref, fwaas-v2: Remove \u0027status\u0027 from request.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/34af1e37bb417b61110695d88db6186a8aece02f"}]}],"author":{"name":"Hongbin Lu","email":"hongbin.lu@huawei.com","date":"2018-06-25 16:37:52.000000000","tz":0},"committer":{"name":"Hongbin Lu","email":"hongbin.lu@huawei.com","date":"2018-08-13 15:45:43.000000000","tz":0},"subject":"Update the segment api extension","message":"Update the segment api extension\n\nThe segment api definition is changed in server side [1]. This patch\nsync up the definition in neutron-lib. In addition, remove the\ndependency from standard-attr-segment on segment since the\ndependency is reserved: we have segment depending on\nstandard-attr-segment with the latest change.\n\n[1] https://review.openstack.org/#/c/571292/\n\nChange-Id: I64780fc435da98d9b5d4e58c9e61d3d0b11d0daf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e3d90fef655b41c36f0e6401a3b19118d496d99c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e3d90fef655b41c36f0e6401a3b19118d496d99c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
