)]}'
{"id":"openstack%2Fneutron-lib~626871","triplet_id":"openstack%2Fneutron-lib~master~I7c16ffb21f88de88ab80a19d91b898be53419514","project":"openstack/neutron-lib","branch":"master","topic":"neutron-interconnection","hashtags":[],"change_id":"I7c16ffb21f88de88ab80a19d91b898be53419514","subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","status":"MERGED","created":"2018-12-21 10:50:52.000000000","updated":"2019-11-15 07:57:16.000000000","submitted":"2019-02-22 15:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":57,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"626871-1550848078946-dce19ce4","meta_rev_id":"73429a419c9817f277c10117261c0738348b0543","_number":626871,"virtual_id_number":626871,"owner":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"date":"2019-02-22 13:38:55.000000000","_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":27546,"name":"Michal Kelner Mishali","email":"mkelnermishal@vmware.com","username":"mkelnermishal"},{"value":0,"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":2,"date":"2019-02-22 15:07:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"all":[{"value":2,"date":"2019-02-22 09:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-02-22 10:30:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-02-22 13:13:14.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-02-22 13:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":27546,"name":"Michal Kelner Mishali","email":"mkelnermishal@vmware.com","username":"mkelnermishal"},{"value":0,"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-02-22 13:13:14.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":27546,"name":"Michal Kelner Mishali","email":"mkelnermishal@vmware.com","username":"mkelnermishal"},{"value":0,"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"value":0,"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":27546,"name":"Michal Kelner Mishali","email":"mkelnermishal@vmware.com","username":"mkelnermishal"},{"value":0,"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27546,"name":"Michal Kelner Mishali","email":"mkelnermishal@vmware.com","username":"mkelnermishal"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-21 16:04:30.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-01-02 16:49:54.000000000","updated_by":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"reviewer":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"state":"REVIEWER"},{"updated":"2019-02-11 11:41:36.000000000","updated_by":{"_account_id":27546,"name":"Michal Kelner Mishali","email":"mkelnermishal@vmware.com","username":"mkelnermishal"},"reviewer":{"_account_id":27546,"name":"Michal Kelner Mishali","email":"mkelnermishal@vmware.com","username":"mkelnermishal"},"state":"REVIEWER"},{"updated":"2019-02-22 09:46:36.000000000","updated_by":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"reviewer":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"state":"REVIEWER"},{"updated":"2019-02-22 10:30:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-02-22 13:13:14.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-02-22 13:30:49.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-02-22 15:07:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"564fed4911dfecfc6f0c4b05f044ceea88c40577","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2018-12-21 10:50:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52cd8cd64dd119efd5283f62cc8568714fbbd8dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 12:19:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/1/check/build-openstack-api-ref/d7b3143/ : FAILURE in 5m 17s\n- openstack-tox-docs http://logs.openstack.org/71/626871/1/check/openstack-tox-docs/a35f799/html/ : SUCCESS in 4m 53s\n- openstack-tox-cover http://logs.openstack.org/71/626871/1/check/openstack-tox-cover/8b6f0a6/ : FAILURE in 4m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/1/check/openstack-tox-lower-constraints/eb2759e/ : FAILURE in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/1/check/openstack-tox-pep8/3df39cb/ : FAILURE in 4m 36s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/1/check/openstack-tox-py27/764013a/ : FAILURE in 4m 37s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/1/check/openstack-tox-py35/8223c16/ : FAILURE in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/1/check/openstack-tox-py36/a69f0de/ : FAILURE in 4m 33s\n- tempest-full http://logs.openstack.org/71/626871/1/check/tempest-full/fe09189/ : SUCCESS in 1h 21m 51s\n- tempest-full-py3 http://logs.openstack.org/71/626871/1/check/tempest-full-py3/ea1e868/ : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1486228bce51e6c0721db761f31c31605491dcbe","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2018-12-21 15:28:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"156db0a3b6175eea7736fea8b6c56ee2d4732031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 16:55:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/2/check/build-openstack-api-ref/dd397f0/ : FAILURE in 4m 37s\n- openstack-tox-docs http://logs.openstack.org/71/626871/2/check/openstack-tox-docs/3b6df57/html/ : SUCCESS in 4m 30s\n- openstack-tox-cover http://logs.openstack.org/71/626871/2/check/openstack-tox-cover/094b5d1/cover/ : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/2/check/openstack-tox-lower-constraints/79ec70d/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/2/check/openstack-tox-pep8/1c9b391/ : SUCCESS in 6m 37s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/2/check/openstack-tox-py27/d4015b5/ : SUCCESS in 4m 59s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/2/check/openstack-tox-py35/6d109bc/ : SUCCESS in 4m 48s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/2/check/openstack-tox-py36/a1ba335/ : SUCCESS in 4m 16s\n- tempest-full http://logs.openstack.org/71/626871/2/check/tempest-full/da8c103/ : SUCCESS in 1h 23m 00s\n- tempest-full-py3 http://logs.openstack.org/71/626871/2/check/tempest-full-py3/7cf8876/ : SUCCESS in 1h 16m 36s","accounts_in_message":[],"_revision_number":2},{"id":"754575363a71f709c603c475ebd93ece87d8427f","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-01-02 16:49:54.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4f1b26367157c3cdb53612f37c721d8f5c93d2e7","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-01-09 09:08:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f3cb7cb39aef57e495011477072616904d355779","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-09 10:49:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/3/check/build-openstack-api-ref/ce924b7/html/ : SUCCESS in 6m 38s\n- openstack-tox-docs http://logs.openstack.org/71/626871/3/check/openstack-tox-docs/4ffb995/html/ : SUCCESS in 3m 40s\n- openstack-tox-cover http://logs.openstack.org/71/626871/3/check/openstack-tox-cover/5165de8/cover/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/3/check/openstack-tox-lower-constraints/4f45987/ : SUCCESS in 5m 34s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/3/check/openstack-tox-pep8/fc7dd16/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/3/check/openstack-tox-py27/9c35820/ : SUCCESS in 5m 40s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/3/check/openstack-tox-py35/acf5671/ : SUCCESS in 4m 25s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/3/check/openstack-tox-py36/34729ef/ : SUCCESS in 4m 26s\n- tempest-full http://logs.openstack.org/71/626871/3/check/tempest-full/787ee9c/ : SUCCESS in 1h 40m 01s\n- tempest-full-py3 http://logs.openstack.org/71/626871/3/check/tempest-full-py3/1970661/ : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":3},{"id":"135c2bfd0d744312423694cb68156a899090aba2","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-01-09 11:29:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"88cc98bace5ac613b992b07d5ca0e5e7509e0e1f","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-01-14 15:43:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"aa10fd207ed538ee0675904c96b72694085829f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-14 17:08:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/4/check/build-openstack-api-ref/a41cd49/html/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/71/626871/4/check/openstack-tox-docs/b683c7d/html/ : SUCCESS in 4m 43s\n- openstack-tox-cover http://logs.openstack.org/71/626871/4/check/openstack-tox-cover/846fedd/cover/ : SUCCESS in 6m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/4/check/openstack-tox-lower-constraints/40f02b0/ : SUCCESS in 5m 45s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/4/check/openstack-tox-pep8/dcf6ba4/ : SUCCESS in 9m 05s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/4/check/openstack-tox-py27/41a6c5a/ : SUCCESS in 8m 07s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/4/check/openstack-tox-py35/7633bd0/ : SUCCESS in 7m 39s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/4/check/openstack-tox-py36/e0b2507/ : SUCCESS in 6m 56s\n- tempest-full http://logs.openstack.org/71/626871/4/check/tempest-full/96a313f/ : SUCCESS in 1h 20m 44s\n- tempest-full-py3 http://logs.openstack.org/71/626871/4/check/tempest-full-py3/878b55b/ : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":4},{"id":"34d28e4a59d5bfda3ae4d8cbdbf6b1574e00808b","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-01-15 13:18:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(10 comments)\n\nAlso:\n- Nit: I assume this is part of a BP, so would be best if the topic was bp/{the BP name}.\n- This will also need a release note.","accounts_in_message":[],"_revision_number":4},{"id":"dc21a1195d769c69b7b1be94307f8ffea3fbf4ad","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-01-24 15:57:18.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f18fa35dca0f657699ac659587a4d7b117d54f31","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-01-24 17:10:24.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2cccbaa7c4dcd2651092e5efc2d63cb9a9e0affd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-24 19:18:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/5/check/build-openstack-api-ref/b873118/html/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/71/626871/5/check/openstack-tox-docs/186151c/html/ : SUCCESS in 5m 02s\n- openstack-tox-cover http://logs.openstack.org/71/626871/5/check/openstack-tox-cover/2467151/cover/ : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/5/check/openstack-tox-lower-constraints/92c315d/ : SUCCESS in 6m 21s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/5/check/openstack-tox-pep8/cf9b040/ : SUCCESS in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/5/check/openstack-tox-py27/8fd9e60/ : SUCCESS in 4m 23s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/5/check/openstack-tox-py35/79b37f7/ : SUCCESS in 5m 45s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/5/check/openstack-tox-py36/9f11984/ : SUCCESS in 4m 45s\n- tempest-full http://logs.openstack.org/71/626871/5/check/tempest-full/77915ff/ : TIMED_OUT in 2h 04m 43s\n- tempest-full-py3 http://logs.openstack.org/71/626871/5/check/tempest-full-py3/c46b19e/ : SUCCESS in 1h 32m 18s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/5/check/build-openstack-releasenotes/6e62e3d/html/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":5},{"id":"4703bb1c57f15ba2a65f636a444f539331ea5a74","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-01-24 21:25:40.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"6235537f32ec8d184187a924aca3a5b5ab60f79f","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-01-24 21:25:54.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d6c3d918d1724320d5d2d858db3a59a4731d3df9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-24 23:34:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/6/check/build-openstack-api-ref/f0a2c05/html/ : SUCCESS in 5m 54s\n- openstack-tox-docs http://logs.openstack.org/71/626871/6/check/openstack-tox-docs/b73f19e/html/ : SUCCESS in 4m 10s\n- openstack-tox-cover http://logs.openstack.org/71/626871/6/check/openstack-tox-cover/bbfd393/cover/ : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/6/check/openstack-tox-lower-constraints/2e87ebd/ : SUCCESS in 5m 10s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/6/check/openstack-tox-pep8/0f28a1f/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/6/check/openstack-tox-py27/ca6f036/ : SUCCESS in 5m 49s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/6/check/openstack-tox-py35/e3ffe6c/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/6/check/openstack-tox-py36/9d4bc08/ : SUCCESS in 4m 37s\n- tempest-full http://logs.openstack.org/71/626871/6/check/tempest-full/bf43d98/ : SUCCESS in 2h 02m 23s\n- tempest-full-py3 http://logs.openstack.org/71/626871/6/check/tempest-full-py3/af38683/ : SUCCESS in 1h 31m 56s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/6/check/build-openstack-releasenotes/352cf1d/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":6},{"id":"3fa1d28c85edd50744cc09f72dda9976bdf509ff","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-01-25 10:25:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fa9995922dcd147155bdbb4c1218d914cd45d29d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-25 12:04:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/7/check/build-openstack-api-ref/f8108dd/html/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/71/626871/7/check/openstack-tox-docs/ed9ccc1/html/ : SUCCESS in 4m 03s\n- openstack-tox-cover http://logs.openstack.org/71/626871/7/check/openstack-tox-cover/16e23b4/cover/ : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/7/check/openstack-tox-lower-constraints/83c69d6/ : SUCCESS in 5m 44s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/7/check/openstack-tox-pep8/c67acd4/ : SUCCESS in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/7/check/openstack-tox-py27/5019520/ : SUCCESS in 5m 41s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/7/check/openstack-tox-py35/4bb1738/ : SUCCESS in 5m 42s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/7/check/openstack-tox-py36/2e9e03e/ : SUCCESS in 5m 59s\n- tempest-full http://logs.openstack.org/71/626871/7/check/tempest-full/2aca612/ : SUCCESS in 1h 33m 31s\n- tempest-full-py3 http://logs.openstack.org/71/626871/7/check/tempest-full-py3/c352790/ : SUCCESS in 1h 36m 43s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/7/check/build-openstack-releasenotes/066bdff/html/ : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":7},{"id":"858ad73b8cac5f2c5d29f6c791e0be94bf5e0cc4","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-02-08 16:43:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1a3645ff46975df1dcafa16e6d5ca06cda7fce3b","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-02-08 20:22:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2479eed4b8bbb7fbbc0beaa789088039cb28b669","author":{"_account_id":27546,"name":"Michal Kelner Mishali","email":"mkelnermishal@vmware.com","username":"mkelnermishal"},"date":"2019-02-11 11:41:36.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"fdc68f74e7ffbb7405fe98a9e7da94284e1a718f","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-11 13:22:25.000000000","message":"Patch Set 7: Code-Review-1\n\n(7 comments)\n\nSmall things and nits.","accounts_in_message":[],"_revision_number":7},{"id":"78e0373149c733db3267b548f60651f42933842e","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-13 15:14:55.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"dc711ee7ebfa3b8022711495f15daa1d83fa9278","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-13 15:15:18.000000000","message":"Patch Set 7:\n\n(8 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9de7d47435d5bdf3cf8bdbe3e4e0b30630a56785","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 17:05:37.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/8/check/build-openstack-api-ref/ce0f543/html/ : SUCCESS in 4m 33s\n- openstack-tox-docs http://logs.openstack.org/71/626871/8/check/openstack-tox-docs/2df708e/html/ : SUCCESS in 3m 52s\n- openstack-tox-cover http://logs.openstack.org/71/626871/8/check/openstack-tox-cover/935d4b6/cover/ : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/8/check/openstack-tox-lower-constraints/a4b54d9/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/8/check/openstack-tox-pep8/58c0485/ : FAILURE in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/8/check/openstack-tox-py27/1ff5c7c/ : SUCCESS in 4m 14s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/8/check/openstack-tox-py35/a2807d3/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/8/check/openstack-tox-py36/87e6052/ : SUCCESS in 5m 08s\n- tempest-full http://logs.openstack.org/71/626871/8/check/tempest-full/5da0e68/ : SUCCESS in 1h 45m 06s\n- tempest-full-py3 http://logs.openstack.org/71/626871/8/check/tempest-full-py3/38d8e25/ : SUCCESS in 1h 39m 55s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/8/check/build-openstack-releasenotes/c33bd16/html/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":8},{"id":"2b4269fefad03fc2ad8fe5500282e172378d09f8","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-13 17:15:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"a957c86e393dba0bcd2b0fa5a13a9a6c15b25a0f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-13 19:14:36.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/9/check/build-openstack-api-ref/9d8717f/html/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/71/626871/9/check/openstack-tox-docs/8501d85/html/ : SUCCESS in 4m 31s\n- openstack-tox-cover http://logs.openstack.org/71/626871/9/check/openstack-tox-cover/3a87cbd/cover/ : SUCCESS in 6m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/9/check/openstack-tox-lower-constraints/7cb2ec9/ : SUCCESS in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/9/check/openstack-tox-pep8/b829dc7/ : FAILURE in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/9/check/openstack-tox-py27/2991d63/ : SUCCESS in 5m 35s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/9/check/openstack-tox-py35/6ae4bfb/ : SUCCESS in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/9/check/openstack-tox-py36/28673f6/ : SUCCESS in 5m 52s\n- tempest-full http://logs.openstack.org/71/626871/9/check/tempest-full/7bd209c/ : SUCCESS in 1h 57m 16s\n- tempest-full-py3 http://logs.openstack.org/71/626871/9/check/tempest-full-py3/5d09add/ : SUCCESS in 1h 21m 31s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/9/check/build-openstack-releasenotes/d79a34d/html/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":9},{"id":"2ec3ad1a95254baa0f400189ed70f1fad34491bd","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-14 07:47:50.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"82c28ecd027c1a3dd741ef94770d257faf25ae48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-14 09:52:00.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/10/check/build-openstack-api-ref/78a1ba3/html/ : SUCCESS in 6m 16s\n- openstack-tox-docs http://logs.openstack.org/71/626871/10/check/openstack-tox-docs/983b473/html/ : SUCCESS in 5m 26s\n- openstack-tox-cover http://logs.openstack.org/71/626871/10/check/openstack-tox-cover/6cc405d/cover/ : SUCCESS in 4m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/10/check/openstack-tox-lower-constraints/7e880af/ : SUCCESS in 4m 35s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/10/check/openstack-tox-pep8/b8c0703/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/10/check/openstack-tox-py27/84c4f66/ : SUCCESS in 3m 56s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/10/check/openstack-tox-py35/870089a/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/10/check/openstack-tox-py36/8441106/ : SUCCESS in 4m 33s\n- tempest-full http://logs.openstack.org/71/626871/10/check/tempest-full/620d1a9/ : SUCCESS in 2h 03m 07s\n- tempest-full-py3 http://logs.openstack.org/71/626871/10/check/tempest-full-py3/0b0866b/ : SUCCESS in 1h 28m 36s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/10/check/build-openstack-releasenotes/0a9164e/html/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":10},{"id":"d2fe8adc74a9e279616e90b5a8e7e54767094a62","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-14 13:15:58.000000000","message":"Patch Set 10: Code-Review-1\n\n(5 comments)\n\n-1 is really for my comment on the l3 extension being required/optional.","accounts_in_message":[],"_revision_number":10},{"id":"b9ffeead1f5f2f282396408b295689eee1ad5958","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-14 15:48:01.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"68522804805a09f46d40640495e6f578614f7b25","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-14 15:48:18.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7863b8707f44cc6e364d6ac2aae474b6d0336645","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-14 17:53:57.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/11/check/build-openstack-api-ref/cb76f3b/html/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/71/626871/11/check/openstack-tox-docs/ad92a35/html/ : SUCCESS in 4m 31s\n- openstack-tox-cover http://logs.openstack.org/71/626871/11/check/openstack-tox-cover/d8241b1/cover/ : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/11/check/openstack-tox-lower-constraints/32ccaa5/ : SUCCESS in 7m 59s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/11/check/openstack-tox-pep8/db460ac/ : SUCCESS in 7m 07s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/11/check/openstack-tox-py27/b3a2d31/ : SUCCESS in 6m 32s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/11/check/openstack-tox-py35/061fef1/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/11/check/openstack-tox-py36/5cae18f/ : SUCCESS in 5m 08s\n- tempest-full http://logs.openstack.org/71/626871/11/check/tempest-full/fae0b56/ : SUCCESS in 1h 57m 55s\n- tempest-full-py3 http://logs.openstack.org/71/626871/11/check/tempest-full-py3/a4664fb/ : SUCCESS in 1h 23m 50s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/11/check/build-openstack-releasenotes/0ec31f8/html/ : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":11},{"id":"578dc2f438620c11326b7a160330cd8a8c3c4b3d","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-15 13:27:11.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"8f647728af721471ef818b40d330fe98ce67a977","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-02-19 08:35:42.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)\n\nI have one more doubt. Except that LGTM :)","accounts_in_message":[],"_revision_number":11},{"id":"df8fe94652d79f526248f30b6becad4c47935af8","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-19 13:15:33.000000000","message":"Patch Set 11: -Code-Review\n\nWait for Slawek\u0027s comments.","accounts_in_message":[],"_revision_number":11},{"id":"c7456ff46fece9dca159162c525e27b2615de2ab","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-02-19 14:00:37.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9992dba18daa75542dd678fdb47be9ac9865fc80","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-19 14:08:20.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"e1c6af0a471b299ea1b19c8fbe63fed619d42685","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-02-19 16:04:46.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fb79fb260a06f8e0c7c988882d47ede054a76971","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-19 16:10:11.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/12/check/build-openstack-api-ref/ef3f67d/html/ : SUCCESS in 7m 28s\n- openstack-tox-docs http://logs.openstack.org/71/626871/12/check/openstack-tox-docs/89e91f8/html/ : SUCCESS in 4m 17s\n- openstack-tox-cover http://logs.openstack.org/71/626871/12/check/openstack-tox-cover/9de381b/cover/ : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/12/check/openstack-tox-lower-constraints/a8e6a3a/ : SUCCESS in 6m 49s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/12/check/openstack-tox-pep8/a503b77/ : SUCCESS in 7m 12s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/12/check/openstack-tox-py27/217a433/ : SUCCESS in 5m 34s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/12/check/openstack-tox-py35/b8ec406/ : SUCCESS in 7m 17s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/12/check/openstack-tox-py36/feb091d/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/71/626871/12/check/openstack-tox-py37/f14ed87/ : SUCCESS in 5m 31s\n- tempest-full http://logs.openstack.org/71/626871/12/check/tempest-full/5462278/ : SUCCESS in 1h 45m 33s\n- tempest-full-py3 http://logs.openstack.org/71/626871/12/check/tempest-full-py3/beb70ec/ : SUCCESS in 1h 50m 13s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/12/check/build-openstack-releasenotes/0d95653/html/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":12},{"id":"18648ec4f2877d02e00d9dafa73e39d225efe103","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-02-20 09:10:00.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"846b1cb4ef7ac5b518ac76a2fa3438df5fc55df4","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-02-21 11:09:45.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a842759e9355af225ba17a1d4d05ef75bdb68a8f","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-21 13:32:20.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"014c1b64a882650f6e3d9cfac6b07cf66ed91a42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-21 15:19:46.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/13/check/build-openstack-api-ref/76e37c9/html/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/71/626871/13/check/openstack-tox-docs/e705377/html/ : SUCCESS in 4m 13s\n- openstack-tox-cover http://logs.openstack.org/71/626871/13/check/openstack-tox-cover/c13610b/cover/ : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/13/check/openstack-tox-lower-constraints/dc28a72/ : SUCCESS in 6m 02s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/13/check/openstack-tox-pep8/07eb97c/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/13/check/openstack-tox-py27/2899fd8/ : SUCCESS in 5m 13s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/13/check/openstack-tox-py35/204af30/ : SUCCESS in 5m 24s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/13/check/openstack-tox-py36/73e0a49/ : SUCCESS in 4m 59s\n- openstack-tox-py37 http://logs.openstack.org/71/626871/13/check/openstack-tox-py37/d866d50/ : SUCCESS in 5m 06s\n- tempest-full http://logs.openstack.org/71/626871/13/check/tempest-full/4347d70/ : FAILURE in 1h 43m 44s\n- tempest-full-py3 http://logs.openstack.org/71/626871/13/check/tempest-full-py3/070695f/ : SUCCESS in 1h 40m 22s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/13/check/build-openstack-releasenotes/e37a52f/html/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":13},{"id":"6b1bc72cd4c1fbce2f293fb8fd638891870f14ad","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-21 15:20:42.000000000","message":"Patch Set 13:\n\nrecheck","accounts_in_message":[],"_revision_number":13},{"id":"b4f11ec24df4392ac8222e5e1157ee4d93d26c61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-21 17:25:57.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/13/check/build-openstack-api-ref/8c7359b/html/ : SUCCESS in 4m 52s\n- openstack-tox-docs http://logs.openstack.org/71/626871/13/check/openstack-tox-docs/492efa5/html/ : SUCCESS in 4m 28s\n- openstack-tox-cover http://logs.openstack.org/71/626871/13/check/openstack-tox-cover/76fb3c5/cover/ : SUCCESS in 5m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/13/check/openstack-tox-lower-constraints/b0c7365/ : SUCCESS in 5m 47s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/13/check/openstack-tox-pep8/2cf33f6/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/13/check/openstack-tox-py27/e24e85e/ : SUCCESS in 4m 51s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/13/check/openstack-tox-py35/145f0b7/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/13/check/openstack-tox-py36/0ad8a73/ : SUCCESS in 5m 11s\n- openstack-tox-py37 http://logs.openstack.org/71/626871/13/check/openstack-tox-py37/8309135/ : SUCCESS in 5m 03s\n- tempest-full http://logs.openstack.org/71/626871/13/check/tempest-full/9d767cf/ : SUCCESS in 1h 36m 43s\n- tempest-full-py3 http://logs.openstack.org/71/626871/13/check/tempest-full-py3/16d79ff/ : SUCCESS in 1h 43m 50s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/13/check/build-openstack-releasenotes/0ffdb47/html/ : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":13},{"id":"e3193cd7eadd02bdcfab208b90267a4ff0cfe49a","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-22 09:30:21.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"0330d7c38209b099624ad9753125a0ef698549ad","author":{"_account_id":12021,"name":"Thomas Morin","email":"tmmorin.orange@gmail.com","username":"tmorin"},"date":"2019-02-22 09:46:36.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"ce58148b83d4b0e120e1949330fd035ac4a13e03","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-02-22 10:30:20.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"d16aae8a3a81831e8c214be86f8efa2271600151","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-02-22 10:30:45.000000000","message":"Patch Set 14:\n\nNow, I think that small follow-up to specs would be maybe useful to, just to have it consistent :)","accounts_in_message":[],"_revision_number":14},{"id":"18ee2530d7ddda1cda5143e846250c0fb1d4f755","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 11:24:12.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/14/check/build-openstack-api-ref/462c411/html/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/71/626871/14/check/openstack-tox-docs/11af087/html/ : SUCCESS in 4m 05s\n- openstack-tox-cover http://logs.openstack.org/71/626871/14/check/openstack-tox-cover/6850ca1/cover/ : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/14/check/openstack-tox-lower-constraints/85c4898/ : SUCCESS in 4m 57s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/14/check/openstack-tox-pep8/2724118/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/14/check/openstack-tox-py27/a9e2118/ : SUCCESS in 4m 33s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/14/check/openstack-tox-py35/6d4d9c1/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/14/check/openstack-tox-py36/278d850/ : SUCCESS in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/71/626871/14/check/openstack-tox-py37/ae9e623/ : SUCCESS in 5m 09s\n- tempest-full http://logs.openstack.org/71/626871/14/check/tempest-full/6ada80d/ : SUCCESS in 1h 47m 26s\n- tempest-full-py3 http://logs.openstack.org/71/626871/14/check/tempest-full-py3/237d305/ : SUCCESS in 1h 32m 33s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/14/check/build-openstack-releasenotes/c1e2c2c/html/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":14},{"id":"66bfd4257209a1294c58ae58580ee9dc01fc4162","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-02-22 13:13:14.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"7340ac6d360b61128c634160159119df9afcb71f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 13:13:26.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"aeccdf1e165730655d2d4d2abd7229d73703af3e","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-02-22 13:30:49.000000000","message":"Patch Set 14: Code-Review+2\n\nThanks for the update. This looks more reasonable.\n\n\u003e Now, I think that small follow-up to specs would be maybe useful to, just to have it consistent :)\n\nYeah, it is nice to have.","accounts_in_message":[],"_revision_number":14},{"id":"03b9d4f285daadee795aaa028671e3cbfd7b05f2","author":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"date":"2019-02-22 13:38:55.000000000","message":"Patch Set 14:\n\nYou can find specs update for Stein in change https://review.openstack.org/#/c/598986/","accounts_in_message":[],"_revision_number":14},{"id":"c608e086295325ab1a9b302ac7510d5e2f324fd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 15:07:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"029b753f7b05570ca5e1c575a9c3573427f832d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-22 15:07:58.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/71/626871/14/gate/build-openstack-api-ref/d5ddfdd/html/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/71/626871/14/gate/openstack-tox-docs/0bdd2af/html/ : SUCCESS in 4m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/626871/14/gate/openstack-tox-lower-constraints/72a5b98/ : SUCCESS in 6m 58s\n- openstack-tox-pep8 http://logs.openstack.org/71/626871/14/gate/openstack-tox-pep8/30b5ddc/ : SUCCESS in 4m 48s\n- openstack-tox-py27 http://logs.openstack.org/71/626871/14/gate/openstack-tox-py27/902bbac/ : SUCCESS in 4m 47s\n- openstack-tox-py35 http://logs.openstack.org/71/626871/14/gate/openstack-tox-py35/551bbef/ : SUCCESS in 4m 41s\n- openstack-tox-py36 http://logs.openstack.org/71/626871/14/gate/openstack-tox-py36/0059fc6/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/71/626871/14/gate/openstack-tox-py37/995297a/ : SUCCESS in 5m 04s\n- tempest-full http://logs.openstack.org/71/626871/14/gate/tempest-full/b924d22/ : SUCCESS in 1h 53m 28s\n- tempest-full-py3 http://logs.openstack.org/71/626871/14/gate/tempest-full-py3/8c9ec92/ : SUCCESS in 1h 49m 08s\n- build-openstack-releasenotes http://logs.openstack.org/71/626871/14/gate/build-openstack-releasenotes/23e67cb/html/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":14},{"id":"6180aaffc110be5ed781a3e70a26ea2effa8d32d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-15 07:57:16.000000000","message":"Created a revert of this change as I186a8f42e1d387bfb31025b0c5c224a86d00cd6e","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"888a28d5982a763d8fc8f8211b3c00f763e8b099","revisions":{"c81e8532721a3e91176c5607fb2fb38345082776":{"kind":"REWORK","_number":1,"created":"2018-12-21 10:50:52.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/1"}}},"commit":{"parents":[{"commit":"152342413beb0cf940708069f21e74c2069699e1","subject":"Merge \"Add traffic control exceptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/152342413beb0cf940708069f21e74c2069699e1"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-12-21 10:46:49.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c81e8532721a3e91176c5607fb2fb38345082776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c81e8532721a3e91176c5607fb2fb38345082776"}]},"branch":"refs/heads/master"},"9732d079739e93bd86c84106817d7b99ff5e39ad":{"kind":"REWORK","_number":2,"created":"2018-12-21 15:28:46.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/2"}}},"commit":{"parents":[{"commit":"152342413beb0cf940708069f21e74c2069699e1","subject":"Merge \"Add traffic control exceptions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/152342413beb0cf940708069f21e74c2069699e1"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-12-21 15:28:11.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9732d079739e93bd86c84106817d7b99ff5e39ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9732d079739e93bd86c84106817d7b99ff5e39ad"}]},"branch":"refs/heads/master"},"c5e6c5fe97470bf13f863e6550eeab387f8b4143":{"kind":"REWORK","_number":3,"created":"2019-01-09 09:08:45.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/3"}}},"commit":{"parents":[{"commit":"3ef70e81a13b3cd8e2f7ea77d4c8bd4dbbd23c5b","subject":"Fix the misspelling of \"interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3ef70e81a13b3cd8e2f7ea77d4c8bd4dbbd23c5b"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-01-09 09:07:42.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5e6c5fe97470bf13f863e6550eeab387f8b4143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c5e6c5fe97470bf13f863e6550eeab387f8b4143"}]},"branch":"refs/heads/master"},"cbd75d1c9e486490e2c8507f6d248215592a847c":{"kind":"REWORK","_number":4,"created":"2019-01-14 15:43:01.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/4"}}},"commit":{"parents":[{"commit":"4e1bc16c2fa9857e13889ee50967eb121d067059","subject":"Merge \"rehome db api orm event listener functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4e1bc16c2fa9857e13889ee50967eb121d067059"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-01-14 15:42:34.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cbd75d1c9e486490e2c8507f6d248215592a847c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cbd75d1c9e486490e2c8507f6d248215592a847c"}]},"branch":"refs/heads/master"},"5c481b3c6392fe6d0f8d7c81607ec23589d7d218":{"kind":"REWORK","_number":5,"created":"2019-01-24 17:10:24.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/5"}}},"commit":{"parents":[{"commit":"0e30615d595db2bfe97941416568e4fe1ea1da3b","subject":"Merge \"Move network segment range types to lib constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0e30615d595db2bfe97941416568e4fe1ea1da3b"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-01-24 17:10:00.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nThis patch partially implements the following spec:\nhttps://specs.openstack.org/openstack/neutron-specs/specs/rocky/neutron-inter.html\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5c481b3c6392fe6d0f8d7c81607ec23589d7d218"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5c481b3c6392fe6d0f8d7c81607ec23589d7d218"}]},"branch":"refs/heads/master"},"875fcda6262fe9e43daa2e9a464ddefc57e93ad3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-01-24 21:25:40.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/6"}}},"commit":{"parents":[{"commit":"872c554c8e9a2faf61894a9728bab64dd471cac6","subject":"Merge \"Remove unnecessary attributes and specs from subnet_onboard API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/872c554c8e9a2faf61894a9728bab64dd471cac6"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","date":"2019-01-24 21:25:40.000000000","tz":0},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nThis patch partially implements the following spec:\nhttps://specs.openstack.org/openstack/neutron-specs/specs/rocky/neutron-inter.html\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/875fcda6262fe9e43daa2e9a464ddefc57e93ad3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/875fcda6262fe9e43daa2e9a464ddefc57e93ad3"}]},"branch":"refs/heads/master"},"871772f6f7ffdee663be19371de2f50fe168eeec":{"kind":"REWORK","_number":7,"created":"2019-01-25 10:25:31.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/7"}}},"commit":{"parents":[{"commit":"84dbf048364d666601712422c9bf33e460d657c4","subject":"Merge \"New agent attribute: resources_synced\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/84dbf048364d666601712422c9bf33e460d657c4"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-01-25 10:25:04.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nThis patch partially implements the following spec:\nhttps://specs.openstack.org/openstack/neutron-specs/specs/rocky/neutron-inter.html\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/871772f6f7ffdee663be19371de2f50fe168eeec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/871772f6f7ffdee663be19371de2f50fe168eeec"}]},"branch":"refs/heads/master"},"231fddec20a7c975c6e3550170e9b674371e0f42":{"kind":"REWORK","_number":8,"created":"2019-02-13 15:14:55.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/8"}}},"commit":{"parents":[{"commit":"e9dd3871f96eed26774a0dfce8c08c0dac5af6b6","subject":"Merge \"rehome trunk related callback resource names\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9dd3871f96eed26774a0dfce8c08c0dac5af6b6"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-13 15:13:15.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nThis patch partially implements the following spec:\nhttps://specs.openstack.org/openstack/neutron-specs/specs/rocky/neutron-inter.html\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/231fddec20a7c975c6e3550170e9b674371e0f42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/231fddec20a7c975c6e3550170e9b674371e0f42"}]},"branch":"refs/heads/master"},"9c8ea26e35e5c74091e29824522413ad0f2530ab":{"kind":"REWORK","_number":9,"created":"2019-02-13 17:15:04.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/9"}}},"commit":{"parents":[{"commit":"3133add2c9197a8fb7c7b43ece89591973ecd58a","subject":"Merge \"Add traffic control constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3133add2c9197a8fb7c7b43ece89591973ecd58a"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-13 17:14:50.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nThis patch partially implements the following spec:\nhttps://specs.openstack.org/openstack/neutron-specs/specs/rocky/neutron-inter.html\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9c8ea26e35e5c74091e29824522413ad0f2530ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9c8ea26e35e5c74091e29824522413ad0f2530ab"}]},"branch":"refs/heads/master"},"02a424823d7dac1a80d4f31864df6ca7033ef601":{"kind":"REWORK","_number":10,"created":"2019-02-14 07:47:50.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/10"}}},"commit":{"parents":[{"commit":"3133add2c9197a8fb7c7b43ece89591973ecd58a","subject":"Merge \"Add traffic control constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3133add2c9197a8fb7c7b43ece89591973ecd58a"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-14 07:47:25.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nThis patch partially implements the following spec:\nhttps://specs.openstack.org/openstack/neutron-specs/specs/rocky/neutron-inter.html\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/02a424823d7dac1a80d4f31864df6ca7033ef601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/02a424823d7dac1a80d4f31864df6ca7033ef601"}]},"branch":"refs/heads/master"},"dde69023466091a30c7da42cc470cec921be84f6":{"kind":"REWORK","_number":11,"created":"2019-02-14 15:48:01.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/11"}}},"commit":{"parents":[{"commit":"3133add2c9197a8fb7c7b43ece89591973ecd58a","subject":"Merge \"Add traffic control constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3133add2c9197a8fb7c7b43ece89591973ecd58a"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-14 15:29:23.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nThis patch partially implements the following spec:\nhttps://specs.openstack.org/openstack/neutron-specs/specs/rocky/neutron-inter.html\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dde69023466091a30c7da42cc470cec921be84f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dde69023466091a30c7da42cc470cec921be84f6"}]},"branch":"refs/heads/master"},"3b4724881122e8db3d1bcc30955a69d6037f2e1d":{"kind":"REWORK","_number":12,"created":"2019-02-19 14:08:20.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/12"}}},"commit":{"parents":[{"commit":"cbf993c8df41640fbda1958dcf199119bc12a3e7","subject":"Merge \"Bump lower version of oslo.db to one that includes jitter argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/cbf993c8df41640fbda1958dcf199119bc12a3e7"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-19 14:07:58.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nThis patch partially implements the following spec:\nhttps://specs.openstack.org/openstack/neutron-specs/specs/rocky/neutron-inter.html\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b4724881122e8db3d1bcc30955a69d6037f2e1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b4724881122e8db3d1bcc30955a69d6037f2e1d"}]},"branch":"refs/heads/master"},"b552f0001fd45b73f1510ce61ddfb38a54015d31":{"kind":"REWORK","_number":13,"created":"2019-02-21 13:32:20.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/13"}}},"commit":{"parents":[{"commit":"044ca8e3e4a07502eb4fb89d1425502750628cba","subject":"Merge \"Remove neutron_lib.db from db profiling projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/044ca8e3e4a07502eb4fb89d1425502750628cba"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-21 13:32:00.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nThis patch partially implements the following spec:\nhttps://specs.openstack.org/openstack/neutron-specs/specs/rocky/neutron-inter.html\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b552f0001fd45b73f1510ce61ddfb38a54015d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b552f0001fd45b73f1510ce61ddfb38a54015d31"}]},"branch":"refs/heads/master"},"888a28d5982a763d8fc8f8211b3c00f763e8b099":{"kind":"REWORK","_number":14,"created":"2019-02-22 09:30:21.000000000","uploader":{"_account_id":19990,"name":"Yannick Thomas","email":"ythomas1.ext@orange.com","username":"ythomas1"},"ref":"refs/changes/71/626871/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/71/626871/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/71/626871/14"}}},"commit":{"parents":[{"commit":"044ca8e3e4a07502eb4fb89d1425502750628cba","subject":"Merge \"Remove neutron_lib.db from db profiling projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/044ca8e3e4a07502eb4fb89d1425502750628cba"}]}],"author":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2018-08-08 13:43:10.000000000","tz":120},"committer":{"name":"ythomas1","email":"ythomas1.ext@orange.com","date":"2019-02-22 09:29:13.000000000","tz":60},"subject":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)","message":"\u0027interconnection\u0027 API extension definition (neutron-interconnection)\n\nAdds definition for \u0027interconnection\u0027 API extension for\nneutron-interconnection project.\n\nThis patch partially implements the following spec:\nhttps://specs.openstack.org/openstack/neutron-specs/specs/rocky/neutron-inter.html\n\nSigned-off-by: Thomas Morin \u003cthomas.morin@orange.com\u003e\nSubmitted on behalf of a third-party: Orange\n\nChange-Id: I7c16ffb21f88de88ab80a19d91b898be53419514\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/888a28d5982a763d8fc8f8211b3c00f763e8b099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/888a28d5982a763d8fc8f8211b3c00f763e8b099"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
