)]}'
{"id":"openstack%2Fneutron-lib~628650","triplet_id":"openstack%2Fneutron-lib~master~I0f018fe963256658682676fcd86520ceae590c75","project":"openstack/neutron-lib","branch":"master","hashtags":[],"change_id":"I0f018fe963256658682676fcd86520ceae590c75","subject":"Update hacking version","status":"MERGED","created":"2019-01-04 17:48:34.000000000","updated":"2019-04-24 11:54:38.000000000","submitted":"2019-04-24 11:54:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"628650-1556106878174-11ac1cc1","meta_rev_id":"dbc6fd0769b5d8493aec3c3d1b942ff8858c18c6","_number":628650,"virtual_id_number":628650,"owner":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"date":"2019-04-23 17:08:28.000000000","_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-04-24 11:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"all":[{"value":1,"date":"2019-04-23 08:10:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-04-23 04:48:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-04-23 08:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-04-23 13:12:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-04-23 13:12:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-14 15:20:51.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2019-04-12 16:12:07.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-04-19 12:25:32.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-04-23 04:48:32.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-04-23 08:01:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-04-23 08:10:06.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-04-23 13:12:28.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-04-24 11:54:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acd955b0fbbfd6378c44d7b4834a895f917b1435","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-01-04 17:48:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4cf57516c4c52abb28d851c8a7772c821510b9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-04 19:35:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/1/check/build-openstack-api-ref/e688fbe/html/ : SUCCESS in 5m 27s\n- requirements-check http://logs.openstack.org/50/628650/1/check/requirements-check/52f26b9/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/50/628650/1/check/openstack-tox-docs/c5d8e91/html/ : SUCCESS in 4m 00s\n- openstack-tox-cover http://logs.openstack.org/50/628650/1/check/openstack-tox-cover/471f59c/cover/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/1/check/openstack-tox-lower-constraints/44c95e8/ : SUCCESS in 5m 55s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/1/check/openstack-tox-pep8/4655f35/ : FAILURE in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/1/check/openstack-tox-py27/301cf9c/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/50/628650/1/check/openstack-tox-py35/0aae28d/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/1/check/openstack-tox-py36/63d69e2/ : SUCCESS in 5m 28s\n- tempest-full http://logs.openstack.org/50/628650/1/check/tempest-full/545aea1/ : SUCCESS in 1h 37m 08s\n- tempest-full-py3 http://logs.openstack.org/50/628650/1/check/tempest-full-py3/df26163/ : FAILURE in 51m 39s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/1/check/build-openstack-releasenotes/8c7d6c9/html/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"66c14d519823cce18d1f805ac9d15d53e93704a1","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-01-04 20:37:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11642822e600694da041573ef66d5f78b39885fa","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2019-03-14 15:20:51.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"448892849bf9c8b949a514d86e9cc7bf1f38787e","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-11 05:52:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"064bea8b0cb11eb27f54e0ae50d8260adbe4bc2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-11 07:34:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/2/check/build-openstack-api-ref/e706492/html/ : SUCCESS in 5m 59s\n- requirements-check http://logs.openstack.org/50/628650/2/check/requirements-check/7aa7a37/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/50/628650/2/check/openstack-tox-docs/14ad068/html/ : SUCCESS in 4m 39s\n- openstack-tox-cover http://logs.openstack.org/50/628650/2/check/openstack-tox-cover/30f0e86/cover/ : SUCCESS in 5m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/2/check/openstack-tox-lower-constraints/2f7af17/ : SUCCESS in 5m 24s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/2/check/openstack-tox-pep8/896bfb1/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/2/check/openstack-tox-py27/a7a9836/ : SUCCESS in 5m 24s\n- openstack-tox-py35 http://logs.openstack.org/50/628650/2/check/openstack-tox-py35/23a615a/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/2/check/openstack-tox-py36/2ed106a/ : SUCCESS in 4m 59s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/2/check/openstack-tox-py37/cc19e23/ : SUCCESS in 5m 37s\n- tempest-full http://logs.openstack.org/50/628650/2/check/tempest-full/7392eda/ : SUCCESS in 1h 39m 17s\n- tempest-full-py3 http://logs.openstack.org/50/628650/2/check/tempest-full-py3/fcd8480/ : SUCCESS in 1h 25m 58s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/2/check/build-openstack-releasenotes/caf1f8e/html/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"9e1c4d8603a15f259be02ac1ac87f5a91e8d5c71","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-12 08:17:56.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0af1906158d745e3235f42a739791e638d4a3fc7","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-12 12:13:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39fd487e7d53c4dccaa8f90479e3c95e6ea7f9b0","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-04-12 16:12:07.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"5037a62e83a068fccd802b4e2b2dfe3f0c0198d9","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-13 14:52:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"773d2162c88c8ad1caaf9a6b34b123afb0622698","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-13 16:51:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/3/check/build-openstack-api-ref/8be0da9/html/ : SUCCESS in 5m 25s\n- requirements-check http://logs.openstack.org/50/628650/3/check/requirements-check/a46cf2d/ : SUCCESS in 3m 30s\n- openstack-tox-docs http://logs.openstack.org/50/628650/3/check/openstack-tox-docs/897a8cd/html/ : SUCCESS in 5m 16s\n- openstack-tox-cover http://logs.openstack.org/50/628650/3/check/openstack-tox-cover/d8bca12/cover/ : SUCCESS in 5m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/3/check/openstack-tox-lower-constraints/1f2104a/ : SUCCESS in 6m 13s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/3/check/openstack-tox-pep8/dea7283/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/3/check/openstack-tox-py27/c477e37/ : SUCCESS in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/50/628650/3/check/openstack-tox-py35/cecfb55/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/3/check/openstack-tox-py36/eb12a31/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/3/check/openstack-tox-py37/35f2dbe/ : SUCCESS in 6m 14s\n- tempest-full http://logs.openstack.org/50/628650/3/check/tempest-full/a62fa31/ : SUCCESS in 1h 53m 18s\n- tempest-full-py3 http://logs.openstack.org/50/628650/3/check/tempest-full-py3/e14f96d/ : SUCCESS in 1h 28m 31s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/3/check/build-openstack-releasenotes/d2e5151/html/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3},{"id":"5e2d3f0a44a7d9cafe15ee4143e18fa58d1243e5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-15 07:54:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3315b2992794b298389cfc7cd9d06c851b70487c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-04-15 16:39:04.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"39e52e42bc8b090d5871e08bbb149ae64d2ff607","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-18 02:39:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b33ab4607681525d56d5e90ce37b913b91553e4e","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-18 02:40:59.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"588072ae8df1187c43619789818f81a41e9a2a63","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-18 02:41:20.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6c89db65226d82ab46a2b44f0cb2c3b4e1cbe502","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-18 21:25:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/5/check/build-openstack-api-ref/bbd0c71/html/ : SUCCESS in 9m 22s\n- requirements-check http://logs.openstack.org/50/628650/5/check/requirements-check/ba1d873/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/50/628650/5/check/openstack-tox-docs/48da880/html/ : SUCCESS in 5m 20s\n- openstack-tox-cover http://logs.openstack.org/50/628650/5/check/openstack-tox-cover/8f16b88/cover/ : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/5/check/openstack-tox-lower-constraints/9e69fab/ : SUCCESS in 7m 49s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/5/check/openstack-tox-pep8/61afa32/ : SUCCESS in 6m 15s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/5/check/openstack-tox-py27/3cc4a7b/ : SUCCESS in 7m 48s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/5/check/openstack-tox-py36/8cc0f08/ : SUCCESS in 5m 51s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/5/check/openstack-tox-py37/6da4d6d/ : SUCCESS in 5m 50s\n- tempest-full http://logs.openstack.org/50/628650/5/check/tempest-full/9582f55/ : SUCCESS in 1h 57m 32s\n- tempest-full-py3 http://logs.openstack.org/50/628650/5/check/tempest-full-py3/71e1cd5/ : SUCCESS in 1h 36m 32s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/5/check/build-openstack-releasenotes/57b3fd9/html/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":5},{"id":"3b0db4ed972d9b1cc29bced77b7576df241def9d","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-19 02:08:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c00096c704c10338c902d5846f0d9da0919dc034","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-04-19 12:25:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9601d36df8f2d5e035369996db7b12252faf254a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-20 07:30:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2f4a1dfcdda001c6ede80ba2672c19fd9bc1c496","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 07:30:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"ba7fdda64f36d5e7056c1c4eb8e87596b6be8c3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 09:42:14.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/5/gate/build-openstack-api-ref/cebea39/html/ : SUCCESS in 5m 40s\n- requirements-check http://logs.openstack.org/50/628650/5/gate/requirements-check/625037b/ : FAILURE in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/50/628650/5/gate/openstack-tox-docs/aaf86d6/html/ : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/5/gate/openstack-tox-lower-constraints/56005d3/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/5/gate/openstack-tox-pep8/b3cd8e3/ : SUCCESS in 5m 08s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/5/gate/openstack-tox-py27/7a378af/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/5/gate/openstack-tox-py36/3a5c999/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/5/gate/openstack-tox-py37/565fe9c/ : SUCCESS in 5m 20s\n- tempest-full http://logs.openstack.org/50/628650/5/gate/tempest-full/38b0c64/ : SUCCESS in 1h 40m 45s\n- tempest-full-py3 http://logs.openstack.org/50/628650/5/gate/tempest-full-py3/10ca5c3/ : SUCCESS in 1h 20m 58s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/5/gate/build-openstack-releasenotes/428ca85/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":5},{"id":"83d0241f24b433eb72c87b264f627e56905a77f9","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-21 12:31:01.000000000","message":"Patch Set 5: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"db2b35bfe7f920b31c3ae085a87ce03337aa0bbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 14:01:35.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/5/check/build-openstack-api-ref/bce8ee3/html/ : SUCCESS in 8m 24s\n- requirements-check http://logs.openstack.org/50/628650/5/check/requirements-check/3eff333/ : FAILURE in 4m 19s\n- openstack-tox-docs http://logs.openstack.org/50/628650/5/check/openstack-tox-docs/03d26b2/html/ : SUCCESS in 7m 31s\n- openstack-tox-cover http://logs.openstack.org/50/628650/5/check/openstack-tox-cover/45ec729/cover/ : SUCCESS in 6m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/5/check/openstack-tox-lower-constraints/8213b72/ : SUCCESS in 4m 44s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/5/check/openstack-tox-pep8/d41cf24/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/5/check/openstack-tox-py27/38e701d/ : SUCCESS in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/5/check/openstack-tox-py36/6b16f01/ : SUCCESS in 6m 10s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/5/check/openstack-tox-py37/5b59645/ : SUCCESS in 4m 48s\n- tempest-full http://logs.openstack.org/50/628650/5/check/tempest-full/05def91/ : SUCCESS in 1h 21m 25s\n- tempest-full-py3 http://logs.openstack.org/50/628650/5/check/tempest-full-py3/63fc2a7/ : SUCCESS in 1h 29m 31s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/5/check/build-openstack-releasenotes/3a4ca60/html/ : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":5},{"id":"92d873830a1b29397bebb9b5c8c325e054dbe771","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-21 19:36:37.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"344137bc6297326c9da0eec652c61c4a2a2560a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 21:23:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/6/check/build-openstack-api-ref/677a50f/html/ : SUCCESS in 4m 26s\n- requirements-check http://logs.openstack.org/50/628650/6/check/requirements-check/58b12b0/ : FAILURE in 2m 50s\n- openstack-tox-docs http://logs.openstack.org/50/628650/6/check/openstack-tox-docs/851fd49/html/ : SUCCESS in 3m 49s\n- openstack-tox-cover http://logs.openstack.org/50/628650/6/check/openstack-tox-cover/3e4d689/cover/ : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/6/check/openstack-tox-lower-constraints/bfd305a/ : SUCCESS in 4m 59s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/6/check/openstack-tox-pep8/c6cf515/ : SUCCESS in 4m 31s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/6/check/openstack-tox-py27/6b478fe/ : SUCCESS in 4m 10s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/6/check/openstack-tox-py36/4466635/ : SUCCESS in 4m 04s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/6/check/openstack-tox-py37/0095d12/ : SUCCESS in 4m 33s\n- tempest-full http://logs.openstack.org/50/628650/6/check/tempest-full/a67bc9e/ : SUCCESS in 1h 41m 30s\n- tempest-full-py3 http://logs.openstack.org/50/628650/6/check/tempest-full-py3/54bd15f/ : SUCCESS in 1h 32m 58s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/6/check/build-openstack-releasenotes/0d5b8d1/html/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":6},{"id":"a05e51e6058e4b5695bda00386716448ce4fea6f","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-04-22 20:22:40.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bc36c2f7c4670ae8ea495fe2a3d60c54a2f18888","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 01:11:45.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"01dc86593ca8d5933f202b6b715bce1ee61f5e0e","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 01:12:29.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"3391166a68c80ee8489effb49d0b8d7bcec42bef","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 01:13:21.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"8ab0d8063bccd6a57eea33d71aa9ec12eadd88c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 03:17:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/8/check/build-openstack-api-ref/3cc9941/html/ : SUCCESS in 5m 22s\n- requirements-check http://logs.openstack.org/50/628650/8/check/requirements-check/19f150e/ : SUCCESS in 3m 36s\n- openstack-tox-docs http://logs.openstack.org/50/628650/8/check/openstack-tox-docs/2ccb7f7/html/ : SUCCESS in 4m 19s\n- openstack-tox-cover http://logs.openstack.org/50/628650/8/check/openstack-tox-cover/dda4d1a/cover/ : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/8/check/openstack-tox-lower-constraints/d27635b/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/8/check/openstack-tox-pep8/e7ac70f/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/8/check/openstack-tox-py27/2b27899/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/8/check/openstack-tox-py36/2726878/ : SUCCESS in 5m 00s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/8/check/openstack-tox-py37/fae7dc6/ : SUCCESS in 4m 52s\n- tempest-full http://logs.openstack.org/50/628650/8/check/tempest-full/7e91b77/ : SUCCESS in 1h 48m 40s\n- tempest-full-py3 http://logs.openstack.org/50/628650/8/check/tempest-full-py3/9ab691b/ : SUCCESS in 1h 57m 18s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/8/check/build-openstack-releasenotes/48afe53/html/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":8},{"id":"c9c6aa003fd579084a184cd5855cf9abba936b8e","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-04-23 04:48:32.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"ad9642aefdfa3f2ba69aaa7ee24c1b4c5b9004bf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-23 08:01:44.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"c980c40d005c16acf7cdf0155568994a6bda75b1","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-23 08:10:06.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"add34049720e7a94cb66354af48031b4bd9530b3","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-04-23 13:12:28.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"d10f8535077bc75ef0765c51cc6ecea35804ddef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 13:12:43.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f48efdd76038a5aa4a98e0be4082834ddb43d19b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 15:39:55.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/8/gate/build-openstack-api-ref/4cee7a0/html/ : SUCCESS in 7m 17s\n- requirements-check http://logs.openstack.org/50/628650/8/gate/requirements-check/68af38a/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/50/628650/8/gate/openstack-tox-docs/67a54a3/html/ : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/8/gate/openstack-tox-lower-constraints/49420a9/ : SUCCESS in 7m 24s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/8/gate/openstack-tox-pep8/2705481/ : SUCCESS in 6m 45s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/8/gate/openstack-tox-py27/36a8250/ : SUCCESS in 6m 40s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/8/gate/openstack-tox-py36/8f986bd/ : SUCCESS in 6m 08s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/8/gate/openstack-tox-py37/a4655b9/ : SUCCESS in 6m 52s\n- tempest-full finger://ze11.openstack.org/8739a676e685456999475f740e576530 : RETRY_LIMIT in 55m 30s\n- tempest-full-py3 finger://ze06.openstack.org/7dd4e566055f464db22c58c1c5eca295 : RETRY_LIMIT in 53m 03s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/8/gate/build-openstack-releasenotes/987cd7a/html/ : SUCCESS in 6m 48s","accounts_in_message":[],"_revision_number":8},{"id":"0c7d340294acf859057781910207bf347af7e7c4","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-04-23 17:08:28.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"c5178a8c651ed4d9687cd7d3eb8239dae0d2e59f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 23:07:51.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/8/check/build-openstack-api-ref/fafb98a/html/ : SUCCESS in 4m 49s\n- requirements-check http://logs.openstack.org/50/628650/8/check/requirements-check/107c4df/ : SUCCESS in 3m 24s\n- openstack-tox-docs http://logs.openstack.org/50/628650/8/check/openstack-tox-docs/cdf62be/html/ : SUCCESS in 4m 18s\n- openstack-tox-cover http://logs.openstack.org/50/628650/8/check/openstack-tox-cover/9626f3e/cover/ : SUCCESS in 4m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/8/check/openstack-tox-lower-constraints/46553a6/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/8/check/openstack-tox-pep8/907f481/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/8/check/openstack-tox-py27/fba7ce3/ : SUCCESS in 4m 59s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/8/check/openstack-tox-py36/9e15121/ : SUCCESS in 4m 42s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/8/check/openstack-tox-py37/2e7864a/ : SUCCESS in 4m 50s\n- tempest-full http://logs.openstack.org/50/628650/8/check/tempest-full/75300f6/ : SUCCESS in 1h 43m 27s\n- tempest-full-py3 http://logs.openstack.org/50/628650/8/check/tempest-full-py3/b85af2d/ : SUCCESS in 1h 53m 29s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/8/check/build-openstack-releasenotes/7f1e7a8/html/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":8},{"id":"38595ed51ee7f876663b9483bbd66f2d6088ebdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 23:08:06.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"6666a49be350101822cf306caff885c19b04135a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 01:10:47.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/8/gate/build-openstack-api-ref/4b4e047/html/ : SUCCESS in 5m 09s\n- requirements-check http://logs.openstack.org/50/628650/8/gate/requirements-check/c479eef/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/50/628650/8/gate/openstack-tox-docs/a83da8d/html/ : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/8/gate/openstack-tox-lower-constraints/5961c85/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/8/gate/openstack-tox-pep8/a183b81/ : SUCCESS in 4m 59s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/8/gate/openstack-tox-py27/90a1b87/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/8/gate/openstack-tox-py36/ad58918/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/8/gate/openstack-tox-py37/df76750/ : SUCCESS in 5m 59s\n- tempest-full http://logs.openstack.org/50/628650/8/gate/tempest-full/0285e04/ : SUCCESS in 1h 47m 48s\n- tempest-full-py3 http://logs.openstack.org/50/628650/8/gate/tempest-full-py3/45a2c59/ : FAILURE in 6m 59s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/8/gate/build-openstack-releasenotes/d8fa902/html/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":8},{"id":"bcbf14cd7468546d3c0772f13e4c0884f6e1e114","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-04-24 08:05:45.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"16242e6b82ab3feb9829a6dd5d62a8ea251c5ea4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 09:58:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/8/check/build-openstack-api-ref/2555b44/html/ : SUCCESS in 5m 39s\n- requirements-check http://logs.openstack.org/50/628650/8/check/requirements-check/f690b67/ : SUCCESS in 4m 31s\n- openstack-tox-docs http://logs.openstack.org/50/628650/8/check/openstack-tox-docs/c1bd986/html/ : SUCCESS in 5m 19s\n- openstack-tox-cover http://logs.openstack.org/50/628650/8/check/openstack-tox-cover/d2025e4/cover/ : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/8/check/openstack-tox-lower-constraints/b4df1aa/ : SUCCESS in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/8/check/openstack-tox-pep8/86d2b98/ : SUCCESS in 4m 30s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/8/check/openstack-tox-py27/fb15c73/ : SUCCESS in 4m 30s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/8/check/openstack-tox-py36/d7e1c7a/ : SUCCESS in 5m 51s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/8/check/openstack-tox-py37/9a3b61c/ : SUCCESS in 5m 19s\n- tempest-full http://logs.openstack.org/50/628650/8/check/tempest-full/802e252/ : SUCCESS in 1h 48m 10s\n- tempest-full-py3 http://logs.openstack.org/50/628650/8/check/tempest-full-py3/f189e63/ : SUCCESS in 1h 30m 44s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/8/check/build-openstack-releasenotes/ef10235/html/ : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":8},{"id":"757c959710ed7423784c8c663ba21b3b9a28721a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 10:00:30.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"b8cb0c35e36cc275c62e649089f40c1504b36873","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 11:54:37.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/50/628650/8/gate/build-openstack-api-ref/d63ab9a/html/ : SUCCESS in 6m 21s\n- requirements-check http://logs.openstack.org/50/628650/8/gate/requirements-check/2942673/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/50/628650/8/gate/openstack-tox-docs/006659a/html/ : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/50/628650/8/gate/openstack-tox-lower-constraints/0306414/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/50/628650/8/gate/openstack-tox-pep8/09afc39/ : SUCCESS in 5m 54s\n- openstack-tox-py27 http://logs.openstack.org/50/628650/8/gate/openstack-tox-py27/45a2962/ : SUCCESS in 4m 55s\n- openstack-tox-py36 http://logs.openstack.org/50/628650/8/gate/openstack-tox-py36/30f337d/ : SUCCESS in 4m 17s\n- openstack-tox-py37 http://logs.openstack.org/50/628650/8/gate/openstack-tox-py37/f2e322b/ : SUCCESS in 5m 46s\n- tempest-full http://logs.openstack.org/50/628650/8/gate/tempest-full/defece1/ : SUCCESS in 1h 45m 58s\n- tempest-full-py3 http://logs.openstack.org/50/628650/8/gate/tempest-full-py3/4942946/ : SUCCESS in 1h 31m 59s\n- build-openstack-releasenotes http://logs.openstack.org/50/628650/8/gate/build-openstack-releasenotes/4362e02/html/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":8},{"id":"dbc6fd0769b5d8493aec3c3d1b942ff8858c18c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 11:54:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"975d6641f4fe069517635d5d4ffb04b10288c0a5","revisions":{"c22ab0f8acf23245d78ba6a9de449be9eeeff1a8":{"kind":"REWORK","_number":1,"created":"2019-01-04 17:48:34.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/50/628650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/628650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/1"}}},"commit":{"parents":[{"commit":"d1d6b4e7bc95aab5af662276a97cf67837f68ff5","subject":"Merge \"Modify the judgment method of CIDR and Add utests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d1d6b4e7bc95aab5af662276a97cf67837f68ff5"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 17:48:17.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 17:48:30.000000000","tz":480},"subject":"Update hacking version to latest","message":"Update hacking version to latest\n\nChange-Id: I0f018fe963256658682676fcd86520ceae590c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c22ab0f8acf23245d78ba6a9de449be9eeeff1a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c22ab0f8acf23245d78ba6a9de449be9eeeff1a8"}]},"branch":"refs/heads/master"},"6f2d3546cebc4b478536e88a6e197fa1230e0104":{"kind":"REWORK","_number":2,"created":"2019-04-11 05:52:50.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/50/628650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/628650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/2"}}},"commit":{"parents":[{"commit":"e599f4a02a2861572ec90771d0b24aed40f0654d","subject":"Merge \"fix WarningsFixture for public consumption\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e599f4a02a2861572ec90771d0b24aed40f0654d"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 17:48:17.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-11 05:52:47.000000000","tz":480},"subject":"Update hacking version","message":"Update hacking version\n\nChange-Id: I0f018fe963256658682676fcd86520ceae590c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6f2d3546cebc4b478536e88a6e197fa1230e0104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6f2d3546cebc4b478536e88a6e197fa1230e0104"}]},"branch":"refs/heads/master"},"38c3295f560a40424fbc5d9bfd42e54c41be7e7b":{"kind":"REWORK","_number":3,"created":"2019-04-13 14:52:30.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/50/628650/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/628650/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/3"}}},"commit":{"parents":[{"commit":"e599f4a02a2861572ec90771d0b24aed40f0654d","subject":"Merge \"fix WarningsFixture for public consumption\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e599f4a02a2861572ec90771d0b24aed40f0654d"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 17:48:17.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-13 14:52:05.000000000","tz":480},"subject":"Update hacking version","message":"Update hacking version\n\nChange-Id: I0f018fe963256658682676fcd86520ceae590c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/38c3295f560a40424fbc5d9bfd42e54c41be7e7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/38c3295f560a40424fbc5d9bfd42e54c41be7e7b"}]},"branch":"refs/heads/master"},"9950534a21364bfddf9e2d9816b8d9909dd73aa7":{"kind":"REWORK","_number":4,"created":"2019-04-18 02:39:40.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/50/628650/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/628650/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/4"}}},"commit":{"parents":[{"commit":"e599f4a02a2861572ec90771d0b24aed40f0654d","subject":"Merge \"fix WarningsFixture for public consumption\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e599f4a02a2861572ec90771d0b24aed40f0654d"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 17:48:17.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-18 02:39:27.000000000","tz":480},"subject":"Update hacking version","message":"Update hacking version\n\nChange-Id: I0f018fe963256658682676fcd86520ceae590c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9950534a21364bfddf9e2d9816b8d9909dd73aa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9950534a21364bfddf9e2d9816b8d9909dd73aa7"}]},"branch":"refs/heads/master"},"a93e43d33470491e1c3963cc8d6f5a1e048ee784":{"kind":"REWORK","_number":5,"created":"2019-04-18 02:41:20.000000000","uploader":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"ref":"refs/changes/50/628650/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/628650/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/5"}}},"commit":{"parents":[{"commit":"e599f4a02a2861572ec90771d0b24aed40f0654d","subject":"Merge \"fix WarningsFixture for public consumption\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e599f4a02a2861572ec90771d0b24aed40f0654d"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 17:48:17.000000000","tz":480},"committer":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-04-18 02:41:10.000000000","tz":480},"subject":"Update hacking version","message":"Update hacking version\n\nChange-Id: I0f018fe963256658682676fcd86520ceae590c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a93e43d33470491e1c3963cc8d6f5a1e048ee784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a93e43d33470491e1c3963cc8d6f5a1e048ee784"}]},"branch":"refs/heads/master"},"87a9aefe2bb021c449d496efe9521703640efe58":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-04-21 19:36:37.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/50/628650/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/628650/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/6"}}},"commit":{"parents":[{"commit":"84dc5259027306c1b69054401b1a3c49ac702d6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/84dc5259027306c1b69054401b1a3c49ac702d6c"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 17:48:17.000000000","tz":480},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-04-21 19:36:37.000000000","tz":0},"subject":"Update hacking version","message":"Update hacking version\n\nChange-Id: I0f018fe963256658682676fcd86520ceae590c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/87a9aefe2bb021c449d496efe9521703640efe58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/87a9aefe2bb021c449d496efe9521703640efe58"}]},"branch":"refs/heads/master"},"2edc0679090b6fa3e7205f7ce5b6704fc1361521":{"kind":"REWORK","_number":7,"created":"2019-04-23 01:12:29.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/50/628650/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/628650/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/7"}}},"commit":{"parents":[{"commit":"84dc5259027306c1b69054401b1a3c49ac702d6c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/84dc5259027306c1b69054401b1a3c49ac702d6c"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 17:48:17.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 01:12:25.000000000","tz":0},"subject":"Update hacking version","message":"Update hacking version\n\nChange-Id: I0f018fe963256658682676fcd86520ceae590c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2edc0679090b6fa3e7205f7ce5b6704fc1361521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2edc0679090b6fa3e7205f7ce5b6704fc1361521"}]},"branch":"refs/heads/master"},"975d6641f4fe069517635d5d4ffb04b10288c0a5":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-04-23 01:13:21.000000000","uploader":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"ref":"refs/changes/50/628650/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/50/628650/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/50/628650/8"}}},"commit":{"parents":[{"commit":"954e9f5bfcf5db3cf1d58bd2d176868ef35282d5","subject":"Merge \"L3 Conntrack Helper Extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/954e9f5bfcf5db3cf1d58bd2d176868ef35282d5"}]}],"author":{"name":"zhulingjie","email":"easyzlj@gmail.com","date":"2019-01-04 17:48:17.000000000","tz":480},"committer":{"name":"caoyuan","email":"cao.yuan@99cloud.net","date":"2019-04-23 01:13:21.000000000","tz":0},"subject":"Update hacking version","message":"Update hacking version\n\nChange-Id: I0f018fe963256658682676fcd86520ceae590c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/975d6641f4fe069517635d5d4ffb04b10288c0a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/975d6641f4fe069517635d5d4ffb04b10288c0a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
