)]}'
{"id":"openstack%2Fneutron-lib~645288","triplet_id":"openstack%2Fneutron-lib~master~I7786516672af795ac7f192504102a7388c6b7d80","project":"openstack/neutron-lib","branch":"master","topic":"bug/1821058","hashtags":[],"change_id":"I7786516672af795ac7f192504102a7388c6b7d80","subject":"Add new vif_details parameters","status":"MERGED","created":"2019-03-21 18:05:09.000000000","updated":"2019-06-24 14:18:25.000000000","submitted":"2019-06-24 14:18:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"645288-1561385905721-eb0806c3","meta_rev_id":"8a2c035385ba403f363359b7891c2b70274dd625","_number":645288,"virtual_id_number":645288,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-06-24 14:18:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2019-06-20 21:50:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-06-21 08:15:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-06-24 12:26:32.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-06-24 12:26:32.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-21 18:05:30.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-03-28 16:08:38.000000000","updated_by":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"reviewer":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"state":"REVIEWER"},{"updated":"2019-04-05 22:43:41.000000000","updated_by":{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},"reviewer":{"_account_id":30156,"name":"Igor D.C.","email":"igor.duarte.cardoso@intel.com","username":"igordc"},"state":"REVIEWER"},{"updated":"2019-06-20 21:50:47.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-06-21 08:15:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-06-24 12:26:32.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-06-24 14:18:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e0bc6ad5ec8a2e74104e28460186675e61aca32","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-03-21 18:05:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c295b08364ea4d54f25e73be0e18be6f8154381b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 19:50:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/645288/1/check/openstack-tox-docs/c743a7b/html/ : SUCCESS in 6m 44s\n- openstack-tox-cover http://logs.openstack.org/88/645288/1/check/openstack-tox-cover/6274c0a/cover/ : SUCCESS in 7m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/645288/1/check/openstack-tox-lower-constraints/1d2a8f9/ : SUCCESS in 9m 45s\n- openstack-tox-pep8 http://logs.openstack.org/88/645288/1/check/openstack-tox-pep8/98a4c7f/ : SUCCESS in 6m 43s\n- openstack-tox-py27 http://logs.openstack.org/88/645288/1/check/openstack-tox-py27/d9bc755/ : SUCCESS in 5m 51s\n- openstack-tox-py35 http://logs.openstack.org/88/645288/1/check/openstack-tox-py35/11ae779/ : SUCCESS in 5m 22s\n- openstack-tox-py36 http://logs.openstack.org/88/645288/1/check/openstack-tox-py36/8ebac21/ : SUCCESS in 5m 34s\n- openstack-tox-py37 http://logs.openstack.org/88/645288/1/check/openstack-tox-py37/0717540/ : SUCCESS in 6m 31s\n- tempest-full http://logs.openstack.org/88/645288/1/check/tempest-full/987a3dc/ : SUCCESS in 1h 38m 09s\n- tempest-full-py3 http://logs.openstack.org/88/645288/1/check/tempest-full-py3/be54701/ : SUCCESS in 1h 33m 33s","accounts_in_message":[],"_revision_number":1},{"id":"082a80d8ff3e7133cdb54466453b0295cc9a2ff0","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-03-25 12:16:25.000000000","message":"Patch Set 1: Code-Review-1\n\nCan you please include a release note too?","accounts_in_message":[],"_revision_number":1},{"id":"0ce166fdeee934e61162b34520c165ec5875964f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-03-26 12:11:39.000000000","message":"Patch Set 1:\n\n\u003e Can you please include a release note too?\n\nDone","accounts_in_message":[],"_revision_number":1},{"id":"7ac31ec6e7aac2e3bc0fe88b4bde6625c49f9734","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-03-26 12:17:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b8cbe935434e6e2adf987ea677758564afa99f4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 14:30:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/645288/2/check/openstack-tox-docs/145845c/html/ : SUCCESS in 4m 32s\n- openstack-tox-cover http://logs.openstack.org/88/645288/2/check/openstack-tox-cover/5ef6cf2/cover/ : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/645288/2/check/openstack-tox-lower-constraints/27c8385/ : SUCCESS in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/88/645288/2/check/openstack-tox-pep8/5eaffeb/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/88/645288/2/check/openstack-tox-py27/bf2e5d1/ : SUCCESS in 7m 57s\n- openstack-tox-py35 http://logs.openstack.org/88/645288/2/check/openstack-tox-py35/97e588f/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/88/645288/2/check/openstack-tox-py36/13fdc81/ : SUCCESS in 6m 14s\n- openstack-tox-py37 http://logs.openstack.org/88/645288/2/check/openstack-tox-py37/8928154/ : SUCCESS in 8m 51s\n- tempest-full http://logs.openstack.org/88/645288/2/check/tempest-full/d1edcf8/ : SUCCESS in 1h 39m 53s\n- tempest-full-py3 http://logs.openstack.org/88/645288/2/check/tempest-full-py3/f1bf4dc/ : SUCCESS in 1h 36m 27s\n- build-openstack-releasenotes http://logs.openstack.org/88/645288/2/check/build-openstack-releasenotes/0feca99/ : FAILURE in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"62a6874a990ff417ecfaa0704cab71d4df1176f0","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-03-26 17:43:09.000000000","message":"Patch Set 2:\n\nfix up the release notes failure and i think it\u0027s fine.\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"f3e456f5fca3e8cc6dd0b0ea585dfa4320e1f1b7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-03-26 20:44:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"29652244c049cbd43d29cbd71a705da08f2cd504","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-26 22:55:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/88/645288/3/check/openstack-tox-docs/d8e4012/html/ : SUCCESS in 4m 54s\n- openstack-tox-cover http://logs.openstack.org/88/645288/3/check/openstack-tox-cover/b022b14/cover/ : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/645288/3/check/openstack-tox-lower-constraints/31276a0/ : SUCCESS in 7m 13s\n- openstack-tox-pep8 http://logs.openstack.org/88/645288/3/check/openstack-tox-pep8/5afb51a/ : SUCCESS in 5m 20s\n- openstack-tox-py27 http://logs.openstack.org/88/645288/3/check/openstack-tox-py27/e00d38e/ : SUCCESS in 5m 19s\n- openstack-tox-py35 http://logs.openstack.org/88/645288/3/check/openstack-tox-py35/bd2a8c7/ : SUCCESS in 7m 17s\n- openstack-tox-py36 http://logs.openstack.org/88/645288/3/check/openstack-tox-py36/e4b2b55/ : SUCCESS in 7m 19s\n- openstack-tox-py37 http://logs.openstack.org/88/645288/3/check/openstack-tox-py37/fcb4efd/ : SUCCESS in 6m 05s\n- tempest-full http://logs.openstack.org/88/645288/3/check/tempest-full/c25fd51/ : SUCCESS in 1h 48m 41s\n- tempest-full-py3 http://logs.openstack.org/88/645288/3/check/tempest-full-py3/b320f2f/ : SUCCESS in 1h 16m 38s\n- build-openstack-releasenotes http://logs.openstack.org/88/645288/3/check/build-openstack-releasenotes/9fd5897/ : FAILURE in 4m 52s","accounts_in_message":[],"_revision_number":3},{"id":"5a8c1558b1846ddf0e5139999da6a13d519fa232","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-03-27 12:04:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6bb763d4b125bd7d6b5e07660adf7a40d5b22d82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 13:46:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/645288/4/check/openstack-tox-docs/f7b504e/html/ : SUCCESS in 4m 49s\n- openstack-tox-cover http://logs.openstack.org/88/645288/4/check/openstack-tox-cover/72d7797/cover/ : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/645288/4/check/openstack-tox-lower-constraints/47000fe/ : SUCCESS in 5m 25s\n- openstack-tox-pep8 http://logs.openstack.org/88/645288/4/check/openstack-tox-pep8/018c5d4/ : SUCCESS in 4m 49s\n- openstack-tox-py27 http://logs.openstack.org/88/645288/4/check/openstack-tox-py27/89f1ddb/ : SUCCESS in 5m 13s\n- openstack-tox-py35 http://logs.openstack.org/88/645288/4/check/openstack-tox-py35/7e4afee/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/88/645288/4/check/openstack-tox-py36/9d44758/ : SUCCESS in 5m 32s\n- openstack-tox-py37 http://logs.openstack.org/88/645288/4/check/openstack-tox-py37/9ceb4bf/ : SUCCESS in 5m 13s\n- tempest-full http://logs.openstack.org/88/645288/4/check/tempest-full/4a96a67/ : SUCCESS in 1h 34m 42s\n- tempest-full-py3 http://logs.openstack.org/88/645288/4/check/tempest-full-py3/06fd313/ : SUCCESS in 1h 33m 30s\n- build-openstack-releasenotes http://logs.openstack.org/88/645288/4/check/build-openstack-releasenotes/e2381ed/html/ : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":4},{"id":"a2e3c7d8866acf30de37cd2579fc38fc6f9fa098","author":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"date":"2019-03-27 16:32:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"cbd5747a47030c5ffbc0f3024b54af3b67c97343","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-03-28 07:25:57.000000000","message":"Patch Set 4: Code-Review-1\n\nLet\u0027s wait with this until spec will be merged, ok?","accounts_in_message":[],"_revision_number":4},{"id":"76d4071fd858605ece91d54d6be1c8ba36a69403","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-03-28 10:44:33.000000000","message":"Patch Set 4: Workflow-1\n\nYes, my fault. I should -W this patch in the first place.","accounts_in_message":[],"_revision_number":4},{"id":"db0f596eed01428901dbfc0569db2d390602fb01","author":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"date":"2019-03-28 16:08:34.000000000","message":"Patch Set 4: Code-Review-1","accounts_in_message":[],"_revision_number":4},{"id":"37984a02911451101328c58430cb23a0c1c8a663","author":{"_account_id":17120,"name":"Manjeet Singh Bhatia","email":"manjeet.s.bhatia@intel.com","username":"manjeets"},"date":"2019-03-28 16:08:38.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"90b288e1f960b1e575ff2533b03edfe088c795a1","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-29 15:44:14.000000000","message":"Patch Set 4:\n\nIs this patch still active?","accounts_in_message":[],"_revision_number":4},{"id":"ee54afcd680c48c69d98c5d320d5624d41db41bb","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-30 06:46:50.000000000","message":"Patch Set 4:\n\n\u003e Is this patch still active?\n\nAFAICT it will be revived by Rodolfo. RFE was approved recently so I hope he will resume work on this soon :)","accounts_in_message":[],"_revision_number":4},{"id":"d38ae4b6cbeed55ec26d5e160c5d76271cbc5d64","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-06-18 14:28:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ed7d0e58f3bdf051a26ec99a56abfe89e25b2316","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-06-18 14:28:40.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ea45bb164ef069dd470821629be96bdebf9f2f08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 16:43:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/645288/6/check/openstack-tox-docs/f4062ff/html/ : SUCCESS in 4m 31s\n- openstack-tox-cover http://logs.openstack.org/88/645288/6/check/openstack-tox-cover/09755cc/cover/ : SUCCESS in 4m 55s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/645288/6/check/openstack-tox-lower-constraints/311266b/ : SUCCESS in 5m 26s\n- openstack-tox-pep8 http://logs.openstack.org/88/645288/6/check/openstack-tox-pep8/100403a/ : SUCCESS in 6m 46s\n- openstack-tox-py27 http://logs.openstack.org/88/645288/6/check/openstack-tox-py27/6be7c24/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/88/645288/6/check/openstack-tox-py36/43aad0f/ : SUCCESS in 5m 15s\n- openstack-tox-py37 http://logs.openstack.org/88/645288/6/check/openstack-tox-py37/c6c19e4/ : SUCCESS in 5m 37s\n- tempest-full http://logs.openstack.org/88/645288/6/check/tempest-full/cd6701e/ : SUCCESS in 1h 50m 43s\n- tempest-full-py3 http://logs.openstack.org/88/645288/6/check/tempest-full-py3/9a92185/ : SUCCESS in 1h 49m 18s\n- build-openstack-releasenotes http://logs.openstack.org/88/645288/6/check/build-openstack-releasenotes/918be79/html/ : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":6},{"id":"a65d7d22d45d73e1b2e6c90e3c624cfa26dbed2f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-20 21:50:47.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"c08139d37239c854e0e8f6c17484fb8d1f5c80bf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-21 08:15:33.000000000","message":"Patch Set 6: Code-Review+2\n\nIt\u0027s good for me but for sake of proper workflow lets first merge spec https://review.opendev.org/#/c/645173/4 and then this patch :)","accounts_in_message":[],"_revision_number":6},{"id":"b0eb6820c3787fe16dafc17dda9cd31a24fd9ff4","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-24 12:26:32.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"07a6df49cff37d3bbf2692fb3c9f57ee4de3ed24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 12:26:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3784f3f764c355cc6c6b5b6966d6e385ef2ccff3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:18:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"e8205ae82570ab3ebbf607b11507d13bbe11ff31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 14:18:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/88/645288/6/gate/openstack-tox-docs/94055da/html/ : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/645288/6/gate/openstack-tox-lower-constraints/a365ca4/ : SUCCESS in 5m 23s\n- openstack-tox-pep8 http://logs.openstack.org/88/645288/6/gate/openstack-tox-pep8/892f430/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/88/645288/6/gate/openstack-tox-py27/5b95966/ : SUCCESS in 4m 42s\n- openstack-tox-py36 http://logs.openstack.org/88/645288/6/gate/openstack-tox-py36/3e446e4/ : SUCCESS in 5m 04s\n- openstack-tox-py37 http://logs.openstack.org/88/645288/6/gate/openstack-tox-py37/01e04fe/ : SUCCESS in 5m 25s\n- tempest-full http://logs.openstack.org/88/645288/6/gate/tempest-full/9ff9102/ : SUCCESS in 1h 46m 46s\n- tempest-full-py3 http://logs.openstack.org/88/645288/6/gate/tempest-full-py3/63ce4ba/ : SUCCESS in 1h 27m 54s\n- build-openstack-releasenotes http://logs.openstack.org/88/645288/6/gate/build-openstack-releasenotes/bd25eaf/html/ : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c26573f847958fc090ba55b9c086a1cc5756b15d","revisions":{"f7a82a4246e7a705cd8e3ff425194e67613ccd3b":{"kind":"REWORK","_number":1,"created":"2019-03-21 18:05:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/645288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/88/645288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/1"}}},"commit":{"parents":[{"commit":"3b1793eb5f43dd4dc235f0c4966531bb341a2341","subject":"Update master for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b1793eb5f43dd4dc235f0c4966531bb341a2341"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-03-21 18:01:46.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-03-21 18:01:46.000000000","tz":0},"subject":"Add new vif_details parameters","message":"Add new vif_details parameters\n\nAdded three new vif_details parameters:\n  - connectivity: informs about the back-end connectivity level (l2,\n    l3 or not specified).\n  - bound_drivers: a dict of binding level to driver name.\n  - network_events: will map the network-* events tracked.\n\nChange-Id: I7786516672af795ac7f192504102a7388c6b7d80\nRelated-Bug: #1821058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7a82a4246e7a705cd8e3ff425194e67613ccd3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f7a82a4246e7a705cd8e3ff425194e67613ccd3b"}]},"branch":"refs/heads/master"},"a544eb137f3cebe7a0858371d14272154024134b":{"kind":"REWORK","_number":2,"created":"2019-03-26 12:17:53.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/645288/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/88/645288/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/2"}}},"commit":{"parents":[{"commit":"3b1793eb5f43dd4dc235f0c4966531bb341a2341","subject":"Update master for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b1793eb5f43dd4dc235f0c4966531bb341a2341"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-03-21 18:01:46.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-03-26 12:17:40.000000000","tz":0},"subject":"Add new vif_details parameters","message":"Add new vif_details parameters\n\nAdded three new vif_details parameters:\n  - connectivity: informs about the back-end connectivity level (l2,\n    l3 or not specified).\n  - bound_drivers: a dict of binding level to driver name.\n  - network_events: will map the network-* events tracked.\n\nChange-Id: I7786516672af795ac7f192504102a7388c6b7d80\nRelated-Bug: #1821058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a544eb137f3cebe7a0858371d14272154024134b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a544eb137f3cebe7a0858371d14272154024134b"}]},"branch":"refs/heads/master"},"b2c26b18a924ffb1feaa630fafb646387729b810":{"kind":"REWORK","_number":3,"created":"2019-03-26 20:44:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/645288/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/88/645288/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/3"}}},"commit":{"parents":[{"commit":"3b1793eb5f43dd4dc235f0c4966531bb341a2341","subject":"Update master for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b1793eb5f43dd4dc235f0c4966531bb341a2341"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-03-21 18:01:46.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-03-26 20:43:58.000000000","tz":0},"subject":"Add new vif_details parameters","message":"Add new vif_details parameters\n\nAdded three new vif_details parameters:\n  - connectivity: informs about the back-end connectivity level (l2,\n    l3 or not specified).\n  - bound_drivers: a dict of binding level to driver name.\n  - network_events: will map the network-* events tracked.\n\nChange-Id: I7786516672af795ac7f192504102a7388c6b7d80\nRelated-Bug: #1821058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b2c26b18a924ffb1feaa630fafb646387729b810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b2c26b18a924ffb1feaa630fafb646387729b810"}]},"branch":"refs/heads/master"},"f1f68b25a0ce9b291f0165c7c29147bb0bf37f0b":{"kind":"REWORK","_number":4,"created":"2019-03-27 12:04:15.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/645288/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/88/645288/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/4"}}},"commit":{"parents":[{"commit":"3b1793eb5f43dd4dc235f0c4966531bb341a2341","subject":"Update master for stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3b1793eb5f43dd4dc235f0c4966531bb341a2341"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-03-21 18:01:46.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-03-27 12:04:01.000000000","tz":0},"subject":"Add new vif_details parameters","message":"Add new vif_details parameters\n\nAdded three new vif_details parameters:\n  - connectivity: informs about the back-end connectivity level (l2,\n    l3 or not specified).\n  - bound_drivers: a dict of binding level to driver name.\n  - network_events: will map the network-* events tracked.\n\nChange-Id: I7786516672af795ac7f192504102a7388c6b7d80\nRelated-Bug: #1821058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f1f68b25a0ce9b291f0165c7c29147bb0bf37f0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f1f68b25a0ce9b291f0165c7c29147bb0bf37f0b"}]},"branch":"refs/heads/master"},"b3359061ffb274544c98e382eafe75ac06f8af6b":{"kind":"REWORK","_number":5,"created":"2019-06-18 14:28:09.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/645288/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/88/645288/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/5"}}},"commit":{"parents":[{"commit":"d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b","subject":"Fix subnet_dns_publish_fixed_ip extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-03-21 18:01:46.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-06-18 14:27:55.000000000","tz":0},"subject":"Add new vif_details parameters","message":"Add new vif_details parameters\n\nAdded three new vif_details parameters:\n  - connectivity: informs about the back-end connectivity level (l2,\n    l3 or not specified).\n  - bound_drivers: a dict of binding level to driver name.\n  - network_events: will map the network-* events tracked.\n\nChange-Id: I7786516672af795ac7f192504102a7388c6b7d80\nRelated-Bug: #1821058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b3359061ffb274544c98e382eafe75ac06f8af6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b3359061ffb274544c98e382eafe75ac06f8af6b"}]},"branch":"refs/heads/master"},"c26573f847958fc090ba55b9c086a1cc5756b15d":{"kind":"REWORK","_number":6,"created":"2019-06-18 14:28:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/88/645288/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/88/645288/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/88/645288/6"}}},"commit":{"parents":[{"commit":"d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b","subject":"Fix subnet_dns_publish_fixed_ip extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-03-21 18:01:46.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2019-06-18 14:28:32.000000000","tz":0},"subject":"Add new vif_details parameters","message":"Add new vif_details parameters\n\nAdded two new vif_details parameters:\n  - connectivity: informs about the back-end connectivity level (l2,\n    l3 or not specified).\n  - bound_drivers: a dict of binding level to driver name.\n\nChange-Id: I7786516672af795ac7f192504102a7388c6b7d80\nRelated-Bug: #1821058\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c26573f847958fc090ba55b9c086a1cc5756b15d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c26573f847958fc090ba55b9c086a1cc5756b15d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
