)]}'
{"id":"openstack%2Fneutron-lib~657456","triplet_id":"openstack%2Fneutron-lib~master~Id1bdd97634be0de52b3ad0697740cc1621c9682a","project":"openstack/neutron-lib","branch":"master","topic":"fix-warnfixture","hashtags":[],"change_id":"Id1bdd97634be0de52b3ad0697740cc1621c9682a","subject":"use once rather than always for filterwarnings","status":"MERGED","created":"2019-05-06 21:13:42.000000000","updated":"2019-05-08 09:28:15.000000000","submitted":"2019-05-08 09:28:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657456-1557307695631-a31b8f89","meta_rev_id":"72a7cf53dd98f0282886349b33b3c334254fc807","_number":657456,"virtual_id_number":657456,"owner":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-06 21:14:25.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-05-08 09:28:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-05-08 07:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-05-07 15:42:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":1,"date":"2019-05-08 07:37:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-07 15:42:21.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-05-08 07:37:54.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-05-08 09:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e05f6a4f5be4b83c6387664f1cda731f9bf9f8f1","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-06 21:13:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b3b457d7834d79c489bd701f19a05b9a2bfc4e6","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-06 21:14:25.000000000","message":"Patch Set 1:\n\nNo consumers are using the WarningsFixture today and since this patch doesn\u0027t change public API I didn\u0027t include a release note. If others think a release note is needed I can add it.","accounts_in_message":[],"_revision_number":1},{"id":"8f49aa7d57dbd2e996317b88c44727861b27b5c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-06 22:54:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/657456/1/check/openstack-tox-docs/2a1c039/html/ : SUCCESS in 4m 12s\n- openstack-tox-cover http://logs.openstack.org/56/657456/1/check/openstack-tox-cover/2401920/cover/ : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/657456/1/check/openstack-tox-lower-constraints/b99bf87/ : SUCCESS in 5m 12s\n- openstack-tox-pep8 http://logs.openstack.org/56/657456/1/check/openstack-tox-pep8/b38df6e/ : SUCCESS in 4m 55s\n- openstack-tox-py27 http://logs.openstack.org/56/657456/1/check/openstack-tox-py27/137d61e/ : SUCCESS in 7m 04s\n- openstack-tox-py36 http://logs.openstack.org/56/657456/1/check/openstack-tox-py36/138fc10/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/56/657456/1/check/openstack-tox-py37/cbc62b3/ : SUCCESS in 5m 15s\n- tempest-full http://logs.openstack.org/56/657456/1/check/tempest-full/15d2c42/ : SUCCESS in 1h 39m 39s\n- tempest-full-py3 http://logs.openstack.org/56/657456/1/check/tempest-full-py3/45ca8cc/ : SUCCESS in 1h 38m 38s","accounts_in_message":[],"_revision_number":1},{"id":"efb15cb80ac00e9ac98f18ec55b568cead0db59c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-07 15:42:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9cd9c5873f8bcf4cf90f75a9e9bf209769a13e3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-08 07:37:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5011ffd585db5afd2b9d3b2bad9e49e7290605b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 07:38:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56b72d2a68a4f35615a05471b8031ce1d9c4f5bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 09:28:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/56/657456/1/gate/openstack-tox-docs/3023eab/html/ : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/657456/1/gate/openstack-tox-lower-constraints/b0b3436/ : SUCCESS in 5m 37s\n- openstack-tox-pep8 http://logs.openstack.org/56/657456/1/gate/openstack-tox-pep8/04a17e7/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/56/657456/1/gate/openstack-tox-py27/2ac4c57/ : SUCCESS in 4m 11s\n- openstack-tox-py36 http://logs.openstack.org/56/657456/1/gate/openstack-tox-py36/03869b5/ : SUCCESS in 5m 22s\n- openstack-tox-py37 http://logs.openstack.org/56/657456/1/gate/openstack-tox-py37/86faa3f/ : SUCCESS in 4m 27s\n- tempest-full http://logs.openstack.org/56/657456/1/gate/tempest-full/ad63dd2/ : SUCCESS in 1h 45m 06s\n- tempest-full-py3 http://logs.openstack.org/56/657456/1/gate/tempest-full-py3/0d9f5d6/ : SUCCESS in 1h 32m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b7bf70b1fec61730d7279ab93e28a3ab62c28d91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-08 09:28:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c220374779ddb73c3d7870fcc0055cfe92a6dc84","revisions":{"c220374779ddb73c3d7870fcc0055cfe92a6dc84":{"kind":"REWORK","_number":1,"created":"2019-05-06 21:13:42.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/56/657456/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/56/657456/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/657456/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/657456/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/56/657456/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/56/657456/1"}}},"commit":{"parents":[{"commit":"978489953b867ed87c2a5cdc31bec17e1af42224","subject":"Merge \"Update hacking version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/978489953b867ed87c2a5cdc31bec17e1af42224"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-06 21:10:23.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-06 21:10:23.000000000","tz":-360},"subject":"use once rather than always for filterwarnings","message":"use once rather than always for filterwarnings\n\nThe previous update to the WarningsFixture in [1] didn\u0027t account for\nthe use of \"once\" vs \"always\" for the filterwarnings call. This patch\nswitches over to \"once\" for parity with neutron.\n\n[1] https://review.opendev.org/#/c/651372/\n\nChange-Id: Id1bdd97634be0de52b3ad0697740cc1621c9682a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c220374779ddb73c3d7870fcc0055cfe92a6dc84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c220374779ddb73c3d7870fcc0055cfe92a6dc84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
