)]}'
{"id":"openstack%2Fneutron-lib~657927","triplet_id":"openstack%2Fneutron-lib~master~I92e07412e06d5e0bcd4a3ef63643d140243f8205","project":"openstack/neutron-lib","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I92e07412e06d5e0bcd4a3ef63643d140243f8205","subject":"Update Python 3 test runtimes for Train","status":"MERGED","created":"2019-05-09 06:26:40.000000000","updated":"2019-05-16 22:49:31.000000000","submitted":"2019-05-16 22:49:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657927-1558046971233-deb1f29f","meta_rev_id":"36a3ac43a832b67b089533cf9a288db7c5e0039d","_number":657927,"virtual_id_number":657927,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":2,"date":"2019-05-16 22:49:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30382,"name":"xuhonghui","email":"xuhh@awcloud.com","username":"xuhh@awcloud.com"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-05-16 15:45:13.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-05-16 21:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30382,"name":"xuhonghui","email":"xuhh@awcloud.com","username":"xuhh@awcloud.com"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":1,"date":"2019-05-16 21:05:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30382,"name":"xuhonghui","email":"xuhh@awcloud.com","username":"xuhh@awcloud.com"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30382,"name":"xuhonghui","email":"xuhh@awcloud.com","username":"xuhh@awcloud.com"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30382,"name":"xuhonghui","email":"xuhh@awcloud.com","username":"xuhh@awcloud.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-09 08:13:07.000000000","updated_by":{"_account_id":30382,"name":"xuhonghui","email":"xuhh@awcloud.com","username":"xuhh@awcloud.com"},"reviewer":{"_account_id":30382,"name":"xuhonghui","email":"xuhh@awcloud.com","username":"xuhh@awcloud.com"},"state":"REVIEWER"},{"updated":"2019-05-10 04:25:33.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-05-16 15:45:13.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-05-16 21:05:38.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-05-16 22:49:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17f4535c2633c694fa53c213a9d4d1755dc90f0e","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-09 06:26:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09a7979888dd82ee21dca0ffede592d69a474634","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 08:11:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/27/657927/1/check/test-release-openstack/03f60cd/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/27/657927/1/check/openstack-tox-docs/0def408/html/ : SUCCESS in 4m 05s\n- openstack-tox-cover http://logs.openstack.org/27/657927/1/check/openstack-tox-cover/0ce9539/cover/ : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/657927/1/check/openstack-tox-lower-constraints/fa145a0/ : SUCCESS in 5m 05s\n- openstack-tox-pep8 http://logs.openstack.org/27/657927/1/check/openstack-tox-pep8/c032db2/ : SUCCESS in 5m 26s\n- openstack-tox-py27 http://logs.openstack.org/27/657927/1/check/openstack-tox-py27/332b470/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/27/657927/1/check/openstack-tox-py36/f64af81/ : SUCCESS in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/27/657927/1/check/openstack-tox-py37/762bffa/ : SUCCESS in 5m 32s\n- tempest-full http://logs.openstack.org/27/657927/1/check/tempest-full/aaa1233/ : SUCCESS in 1h 38m 36s\n- tempest-full-py3 http://logs.openstack.org/27/657927/1/check/tempest-full-py3/9e061a7/ : SUCCESS in 1h 38m 09s\n- build-openstack-releasenotes http://logs.openstack.org/27/657927/1/check/build-openstack-releasenotes/2524e53/html/ : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2ba75f7719373a6ed4c34ab42d05b5cfca9afe8b","author":{"_account_id":30382,"name":"xuhonghui","email":"xuhh@awcloud.com","username":"xuhh@awcloud.com"},"date":"2019-05-09 08:13:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"614d3037d0ed67d8763c3b28cdf7c5b8375ddd41","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-09 09:36:05.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1ba7ea4e94dca366ef5e541e500b1a3822144e27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-09 11:19:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/27/657927/2/check/test-release-openstack/746af91/ : SUCCESS in 4m 04s\n- openstack-tox-docs http://logs.openstack.org/27/657927/2/check/openstack-tox-docs/cf9a29b/html/ : SUCCESS in 4m 23s\n- openstack-tox-cover http://logs.openstack.org/27/657927/2/check/openstack-tox-cover/262f526/cover/ : SUCCESS in 8m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/657927/2/check/openstack-tox-lower-constraints/423aa96/ : SUCCESS in 5m 42s\n- openstack-tox-pep8 http://logs.openstack.org/27/657927/2/check/openstack-tox-pep8/dd26e2c/ : FAILURE in 5m 11s\n- openstack-tox-py27 http://logs.openstack.org/27/657927/2/check/openstack-tox-py27/d4db753/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/27/657927/2/check/openstack-tox-py36/ff4ce1f/ : SUCCESS in 5m 46s\n- openstack-tox-py37 http://logs.openstack.org/27/657927/2/check/openstack-tox-py37/1214018/ : SUCCESS in 5m 04s\n- tempest-full http://logs.openstack.org/27/657927/2/check/tempest-full/b7eb1da/ : FAILURE in 1h 21m 30s\n- tempest-full-py3 http://logs.openstack.org/27/657927/2/check/tempest-full-py3/2c246fe/ : SUCCESS in 1h 37m 45s\n- build-openstack-releasenotes http://logs.openstack.org/27/657927/2/check/build-openstack-releasenotes/b64e627/html/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":2},{"id":"44b4d377f26d32d97a98868bd2891bda4d644ea2","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-05-10 00:27:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ba03d33f408a3da28a904afa4d36d8864c5fd313","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 03:23:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack http://logs.openstack.org/27/657927/2/check/test-release-openstack/025166e/ : SUCCESS in 3m 46s\n- openstack-tox-docs http://logs.openstack.org/27/657927/2/check/openstack-tox-docs/5110254/html/ : SUCCESS in 4m 22s\n- openstack-tox-cover http://logs.openstack.org/27/657927/2/check/openstack-tox-cover/ad78996/cover/ : SUCCESS in 5m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/657927/2/check/openstack-tox-lower-constraints/af2b33f/ : SUCCESS in 5m 41s\n- openstack-tox-pep8 http://logs.openstack.org/27/657927/2/check/openstack-tox-pep8/963f2e6/ : FAILURE in 5m 50s\n- openstack-tox-py27 http://logs.openstack.org/27/657927/2/check/openstack-tox-py27/beecc6d/ : SUCCESS in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/27/657927/2/check/openstack-tox-py36/67badf2/ : SUCCESS in 5m 53s\n- openstack-tox-py37 http://logs.openstack.org/27/657927/2/check/openstack-tox-py37/896aa96/ : SUCCESS in 5m 09s\n- tempest-full http://logs.openstack.org/27/657927/2/check/tempest-full/2832171/ : SUCCESS in 1h 45m 00s\n- tempest-full-py3 http://logs.openstack.org/27/657927/2/check/tempest-full-py3/a5acc2b/ : SUCCESS in 2h 36m 27s\n- build-openstack-releasenotes http://logs.openstack.org/27/657927/2/check/build-openstack-releasenotes/1d124c7/html/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"ab5ce1078650595907eda64e0be53739c3fe96e1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-05-10 04:25:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dcb381b8398978345d6430c296b86605a3616cd7","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-15 20:22:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a2f9a08a0eab7f152a6fe208be3be9105bbdb773","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-15 20:23:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7fd7ae56b8eb6c296eb949b22cd8d2701afd83b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 22:03:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/27/657927/3/check/test-release-openstack/ef930bc/ : SUCCESS in 3m 40s\n- openstack-tox-docs http://logs.openstack.org/27/657927/3/check/openstack-tox-docs/79c27a6/html/ : SUCCESS in 6m 11s\n- openstack-tox-cover http://logs.openstack.org/27/657927/3/check/openstack-tox-cover/f0d63b8/cover/ : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/657927/3/check/openstack-tox-lower-constraints/fb03d14/ : SUCCESS in 8m 08s\n- openstack-tox-pep8 http://logs.openstack.org/27/657927/3/check/openstack-tox-pep8/6247c5b/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/27/657927/3/check/openstack-tox-py27/828b47a/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/27/657927/3/check/openstack-tox-py36/2f055c2/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/27/657927/3/check/openstack-tox-py37/5067415/ : SUCCESS in 5m 11s\n- tempest-full http://logs.openstack.org/27/657927/3/check/tempest-full/a019cae/ : SUCCESS in 1h 39m 36s\n- tempest-full-py3 http://logs.openstack.org/27/657927/3/check/tempest-full-py3/00cb88f/ : SUCCESS in 1h 26m 07s\n- build-openstack-releasenotes http://logs.openstack.org/27/657927/3/check/build-openstack-releasenotes/06c2fbe/html/ : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"6755ffb5f564869c63814e1a31b7d73c5c189961","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-16 15:45:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"adc4d64c807108adebe42dfd55e77e7ed43eb666","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-05-16 21:05:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5188d2690f2d557c1b52974ad0b5efc6cceca039","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 21:05:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6e25daf86c079bf53721e132ee96f26252f7c4cf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:49:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"294ebba8c28a1a3a8379f1581831c06163eeb51b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 22:49:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/27/657927/3/gate/test-release-openstack/cf11371/ : SUCCESS in 3m 31s\n- openstack-tox-docs http://logs.openstack.org/27/657927/3/gate/openstack-tox-docs/627763a/html/ : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/657927/3/gate/openstack-tox-lower-constraints/4a04bcb/ : SUCCESS in 5m 43s\n- openstack-tox-pep8 http://logs.openstack.org/27/657927/3/gate/openstack-tox-pep8/4bcfff0/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/27/657927/3/gate/openstack-tox-py27/6f7294d/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/27/657927/3/gate/openstack-tox-py36/ecc9b2d/ : SUCCESS in 4m 58s\n- openstack-tox-py37 http://logs.openstack.org/27/657927/3/gate/openstack-tox-py37/fd31ed7/ : SUCCESS in 4m 38s\n- tempest-full http://logs.openstack.org/27/657927/3/gate/tempest-full/00d85f4/ : SUCCESS in 1h 29m 05s\n- tempest-full-py3 http://logs.openstack.org/27/657927/3/gate/tempest-full-py3/88311a7/ : SUCCESS in 1h 38m 32s\n- build-openstack-releasenotes http://logs.openstack.org/27/657927/3/gate/build-openstack-releasenotes/57514e4/html/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"57d7d3baa116f172d361d14075f5f16143ddec7b","revisions":{"dd396e2a6dbfc9c25820e363ebab3e55e1cf3653":{"kind":"REWORK","_number":1,"created":"2019-05-09 06:26:40.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/27/657927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/27/657927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/1"}}},"commit":{"parents":[{"commit":"bcd898220ff53b3fed46cef8c460269dd6af3492","subject":"Merge \"Use AssociationProxyInstance instead of AssociationProxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bcd898220ff53b3fed46cef8c460269dd6af3492"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:01.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:01.000000000","tz":480},"subject":"Update Python 3 test runtimes for Train","message":"Update Python 3 test runtimes for Train\n\nFollowing the process set down in the TC resolution on updating Python\nversions:\n\nhttps://governance.openstack.org/tc/resolutions/20181024-python-update-process.html\n\nChange-Id: I92e07412e06d5e0bcd4a3ef63643d140243f8205\nDepends-On: https://review.opendev.org/#/c/641878/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dd396e2a6dbfc9c25820e363ebab3e55e1cf3653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dd396e2a6dbfc9c25820e363ebab3e55e1cf3653"}]},"branch":"refs/heads/master"},"c907ba09634d9f9d9a5ba70ce9bf969afc239738":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-09 09:36:05.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/27/657927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/27/657927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/2"}}},"commit":{"parents":[{"commit":"bcd898220ff53b3fed46cef8c460269dd6af3492","subject":"Merge \"Use AssociationProxyInstance instead of AssociationProxy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bcd898220ff53b3fed46cef8c460269dd6af3492"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:01.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 09:35:12.000000000","tz":480},"subject":"Update Python 3 test runtimes for Train","message":"Update Python 3 test runtimes for Train\n\nThis goal is to implement the process set out in the 2018-10-24 Python\nUpdate Process TC resolution[1], for the Train cycle to ensure unit\ntesting is in place for all of the Tested Runtimes for Train[2].\nIn practice, this generally means adding unit tests for Python 3.7 and dropping\nunit tests for Python 3.5. Using the Zuul template for Train will ensure that\nall projects that support Python3 will be tested against the agreed runtime\nversions, and make it easier to update them in future.\n\n[1]https://governance.openstack.org/tc/resolutions/20181024-python-update-process.html\n[2]https://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I92e07412e06d5e0bcd4a3ef63643d140243f8205\nDepends-On: https://review.opendev.org/#/c/641878/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c907ba09634d9f9d9a5ba70ce9bf969afc239738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c907ba09634d9f9d9a5ba70ce9bf969afc239738"}]},"branch":"refs/heads/master"},"57d7d3baa116f172d361d14075f5f16143ddec7b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2019-05-15 20:22:41.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/27/657927/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/27/657927/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/27/657927/3"}}},"commit":{"parents":[{"commit":"65264a936a6c9d0a8c4eae69acaee2c3ee54b5d6","subject":"Add sort_keys retrieving function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/65264a936a6c9d0a8c4eae69acaee2c3ee54b5d6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-05-09 06:26:01.000000000","tz":480},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-05-15 20:20:36.000000000","tz":-360},"subject":"Update Python 3 test runtimes for Train","message":"Update Python 3 test runtimes for Train\n\nThis goal is to implement the process set out in the 2018-10-24 Python\nUpdate Process TC resolution[1], for the Train cycle to ensure unit\ntesting is in place for all of the Tested Runtimes for Train[2].\nIn practice, this generally means adding unit tests for Python 3.7 and\ndropping unit tests for Python 3.5. Using the Zuul template for Train\nwill ensure that all projects that support Python3 will be tested\nagainst the agreed runtime versions, and make it easier to update them\nin future.\n\nSpecifically this patch:\n- Uses the openstack-python3-train-jobs zuul template rather than\ntemplates for python 36/37.\n- Adds Python 3.7 to the setup.cfg classifier.\n- Adds py37 to the envlist of tox.ini.\n\n[1]https://governance.openstack.org/tc/resolutions/20181024-python-update-process.html\n[2]https://governance.openstack.org/tc/reference/runtimes/train.html\n\nChange-Id: I92e07412e06d5e0bcd4a3ef63643d140243f8205\nDepends-On: https://review.opendev.org/#/c/641878/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/57d7d3baa116f172d361d14075f5f16143ddec7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/57d7d3baa116f172d361d14075f5f16143ddec7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
