)]}'
{"id":"openstack%2Fneutron-lib~659524","triplet_id":"openstack%2Fneutron-lib~master~I8b8877fcee3d2d36c8905c2c35f33cddb172553d","project":"openstack/neutron-lib","branch":"master","topic":"bug/1829332","hashtags":[],"change_id":"I8b8877fcee3d2d36c8905c2c35f33cddb172553d","subject":"Optimize the code that fixes the race condition of DHCP agent.","status":"ABANDONED","created":"2019-05-16 10:37:47.000000000","updated":"2019-05-22 12:04:57.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f22d2020ae6884c8638f42bac6b84c90a943dd89","_number":659524,"virtual_id_number":659524,"owner":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":1,"date":"2019-05-16 12:25:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":-1,"date":"2019-05-16 15:42:45.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":-1,"date":"2019-05-16 19:27:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 12:25:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-16 15:42:45.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-05-16 19:27:33.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"f0f3f347b36b84b60826c5d827fc72eaec7a16fb","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2019-05-16 10:37:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c2ff37742861ec5620f866ac06535569efd3f6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 12:25:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/24/659524/1/check/openstack-tox-docs/a38dc47/html/ : SUCCESS in 4m 30s\n- openstack-tox-cover http://logs.openstack.org/24/659524/1/check/openstack-tox-cover/7a2f285/cover/ : SUCCESS in 6m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/24/659524/1/check/openstack-tox-lower-constraints/195d82c/ : SUCCESS in 6m 30s\n- openstack-tox-pep8 http://logs.openstack.org/24/659524/1/check/openstack-tox-pep8/947f893/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/24/659524/1/check/openstack-tox-py27/14adf4e/ : SUCCESS in 5m 38s\n- openstack-tox-py36 http://logs.openstack.org/24/659524/1/check/openstack-tox-py36/4056eeb/ : SUCCESS in 4m 34s\n- openstack-tox-py37 http://logs.openstack.org/24/659524/1/check/openstack-tox-py37/17e5c69/ : SUCCESS in 6m 40s\n- tempest-full http://logs.openstack.org/24/659524/1/check/tempest-full/675b71f/ : SUCCESS in 1h 41m 59s\n- tempest-full-py3 http://logs.openstack.org/24/659524/1/check/tempest-full-py3/46ea65c/ : SUCCESS in 1h 23m 09s","accounts_in_message":[],"_revision_number":1},{"id":"232984c3e3dddff7f696bccb5c1fdfccb8e956bf","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-16 15:42:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nCan we please also include a release note, probably by adding to the bugs section to note 1829332","accounts_in_message":[],"_revision_number":1},{"id":"44a7b06c258f89c12fe7d6ce3080089f74807628","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-05-16 19:27:33.000000000","message":"Patch Set 1: Code-Review-1\n\nLet\u0027s not approve this until we agree the change in neutron is required, after which we can re-home the new exception.  Should also check in codesearch if this is used elsewhere.","accounts_in_message":[],"_revision_number":1},{"id":"15456066ed895785364df95732b2222e8f284e3b","author":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"date":"2019-05-22 12:04:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a714619181c33a8f7f57b1572f309af907ef62a6","revisions":{"a714619181c33a8f7f57b1572f309af907ef62a6":{"kind":"REWORK","_number":1,"created":"2019-05-16 10:37:47.000000000","uploader":{"_account_id":28329,"name":"yangjianfeng","display_name":"JeffYang","email":"yjf1970231893@gmail.com","username":"yangjianfeng"},"ref":"refs/changes/24/659524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/24/659524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/659524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/659524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/24/659524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/24/659524/1"}}},"commit":{"parents":[{"commit":"65264a936a6c9d0a8c4eae69acaee2c3ee54b5d6","subject":"Add sort_keys retrieving function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/65264a936a6c9d0a8c4eae69acaee2c3ee54b5d6"}]}],"author":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2019-05-16 10:34:35.000000000","tz":480},"committer":{"name":"Yang JianFeng","email":"yjf1970231893@gmail.com","date":"2019-05-16 10:37:21.000000000","tz":480},"subject":"Optimize the code that fixes the race condition of DHCP agent.","message":"Optimize the code that fixes the race condition of DHCP agent.\n\nNeeded-By: https://review.opendev.org/#/c/659523/\n\nChange-Id: I8b8877fcee3d2d36c8905c2c35f33cddb172553d\nCloses-Bug: #1829332\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a714619181c33a8f7f57b1572f309af907ef62a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/a714619181c33a8f7f57b1572f309af907ef62a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
