)]}'
{"id":"openstack%2Fneutron-lib~661938","triplet_id":"openstack%2Fneutron-lib~master~I52c115acdb1a8c464f5fc6273f6d8c4d0c6526e9","project":"openstack/neutron-lib","branch":"master","topic":"bug/1831404","hashtags":[],"change_id":"I52c115acdb1a8c464f5fc6273f6d8c4d0c6526e9","subject":"Add rarp protocol for neutron constants","status":"MERGED","created":"2019-05-29 10:50:45.000000000","updated":"2019-06-14 18:18:23.000000000","submitted":"2019-06-14 18:18:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661938-1560536303075-018dc456","meta_rev_id":"15dfd1caf077481a068e1aff731930c7924649fa","_number":661938,"virtual_id_number":661938,"owner":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":2,"date":"2019-06-14 18:18:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-06-14 09:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-06-14 16:32:37.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":1,"date":"2019-06-14 11:45:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":1,"date":"2019-06-14 14:38:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-14 14:57:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-06-14 16:32:37.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-31 14:41:37.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2019-06-03 21:43:53.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-06-05 19:21:40.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-06-12 11:07:55.000000000","updated_by":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"reviewer":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"state":"REVIEWER"},{"updated":"2019-06-14 09:32:14.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-06-14 11:45:55.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-06-14 14:38:10.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-06-14 14:57:41.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-06-14 16:32:37.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-06-14 18:18:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"374d4990323f76d1d397f0f2e33321d4c52562dc","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-05-29 10:50:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c771c880c5438bf89952690389ca8e11a6eca4c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 13:13:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/661938/1/check/openstack-tox-docs/465df61/html/ : SUCCESS in 4m 32s\n- openstack-tox-cover http://logs.openstack.org/38/661938/1/check/openstack-tox-cover/82da54a/cover/ : SUCCESS in 5m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661938/1/check/openstack-tox-lower-constraints/6f1a587/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/38/661938/1/check/openstack-tox-pep8/4d16025/ : SUCCESS in 5m 42s\n- openstack-tox-py27 http://logs.openstack.org/38/661938/1/check/openstack-tox-py27/d42df3a/ : SUCCESS in 4m 43s\n- openstack-tox-py36 http://logs.openstack.org/38/661938/1/check/openstack-tox-py36/b94ed34/ : SUCCESS in 4m 38s\n- openstack-tox-py37 http://logs.openstack.org/38/661938/1/check/openstack-tox-py37/9b70dd9/ : SUCCESS in 4m 59s\n- tempest-full http://logs.openstack.org/38/661938/1/check/tempest-full/8a46478/ : SUCCESS in 2h 14m 26s\n- tempest-full-py3 http://logs.openstack.org/38/661938/1/check/tempest-full-py3/67b43b3/ : SUCCESS in 1h 44m 44s","accounts_in_message":[],"_revision_number":1},{"id":"68e1f13bcba0a4e8b89d9d70bad8fba6034f39ca","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-05-29 15:33:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nPlease include a release note as well.","accounts_in_message":[],"_revision_number":1},{"id":"1745f62e50f6a043e8ba5619ebe57b961431940c","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-05-31 02:41:31.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fb462143e4441856e6f054148c87b1b5c9857e0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 04:32:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/661938/2/check/openstack-tox-docs/83f2fc2/html/ : SUCCESS in 4m 17s\n- openstack-tox-cover http://logs.openstack.org/38/661938/2/check/openstack-tox-cover/d44b015/cover/ : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661938/2/check/openstack-tox-lower-constraints/1aa3268/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/38/661938/2/check/openstack-tox-pep8/5fa89f7/ : SUCCESS in 4m 44s\n- openstack-tox-py27 http://logs.openstack.org/38/661938/2/check/openstack-tox-py27/922d494/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/38/661938/2/check/openstack-tox-py36/acdf311/ : SUCCESS in 4m 45s\n- openstack-tox-py37 http://logs.openstack.org/38/661938/2/check/openstack-tox-py37/6d070c8/ : SUCCESS in 5m 25s\n- tempest-full http://logs.openstack.org/38/661938/2/check/tempest-full/f537aee/ : SUCCESS in 1h 49m 49s\n- tempest-full-py3 http://logs.openstack.org/38/661938/2/check/tempest-full-py3/8419a49/ : SUCCESS in 1h 37m 41s","accounts_in_message":[],"_revision_number":2},{"id":"6d3f8270c578e4e87baee0ad2fa42f63f6367b92","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-05-31 11:26:44.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d16b09058b6f8fe0c80a31995e3cbca519415572","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 13:33:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/661938/3/check/openstack-tox-docs/b7ff338/html/ : SUCCESS in 4m 16s\n- openstack-tox-cover http://logs.openstack.org/38/661938/3/check/openstack-tox-cover/e28af0b/cover/ : SUCCESS in 4m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661938/3/check/openstack-tox-lower-constraints/78e0e33/ : SUCCESS in 5m 04s\n- openstack-tox-pep8 http://logs.openstack.org/38/661938/3/check/openstack-tox-pep8/6102e74/ : SUCCESS in 5m 59s\n- openstack-tox-py27 http://logs.openstack.org/38/661938/3/check/openstack-tox-py27/b77911b/ : SUCCESS in 4m 50s\n- openstack-tox-py36 http://logs.openstack.org/38/661938/3/check/openstack-tox-py36/353fad3/ : SUCCESS in 4m 27s\n- openstack-tox-py37 http://logs.openstack.org/38/661938/3/check/openstack-tox-py37/3c1d24a/ : SUCCESS in 7m 07s\n- tempest-full http://logs.openstack.org/38/661938/3/check/tempest-full/448ee04/ : SUCCESS in 2h 06m 36s\n- tempest-full-py3 http://logs.openstack.org/38/661938/3/check/tempest-full-py3/20ed550/ : SUCCESS in 1h 37m 11s","accounts_in_message":[],"_revision_number":3},{"id":"e4009b3ed8d8971e9d5519af30a3bd790135cf39","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2019-05-31 14:41:37.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e0c6196592277578307f0866d9b95f9e63bf28ab","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-06-03 02:27:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c26f6cddf46e41a3fd01eaf72978943305bfd624","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-06-03 02:35:21.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"0fcad37a310d4f2d5e62a1819159d5da43118666","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 04:34:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/661938/4/check/openstack-tox-docs/cf6830f/html/ : SUCCESS in 3m 55s\n- openstack-tox-cover http://logs.openstack.org/38/661938/4/check/openstack-tox-cover/89cc873/cover/ : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661938/4/check/openstack-tox-lower-constraints/51c8e11/ : SUCCESS in 4m 54s\n- openstack-tox-pep8 http://logs.openstack.org/38/661938/4/check/openstack-tox-pep8/f8fb2b6/ : SUCCESS in 4m 29s\n- openstack-tox-py27 http://logs.openstack.org/38/661938/4/check/openstack-tox-py27/b35b5e4/ : SUCCESS in 4m 19s\n- openstack-tox-py36 http://logs.openstack.org/38/661938/4/check/openstack-tox-py36/0b6808e/ : SUCCESS in 4m 35s\n- openstack-tox-py37 http://logs.openstack.org/38/661938/4/check/openstack-tox-py37/90de483/ : SUCCESS in 4m 51s\n- tempest-full http://logs.openstack.org/38/661938/4/check/tempest-full/3876cc0/ : SUCCESS in 1h 53m 49s\n- tempest-full-py3 http://logs.openstack.org/38/661938/4/check/tempest-full-py3/a1dc62a/ : SUCCESS in 1h 23m 28s","accounts_in_message":[],"_revision_number":4},{"id":"b13b3e25f717dbc71d8a3a210dcdeb2fb70db2f6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-03 07:40:43.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5830825824ad2f8c5999669fdc77b6015633dee3","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2019-06-03 08:33:36.000000000","message":"Topic changed from bug/1815989 to bug/1831404","accounts_in_message":[],"_revision_number":4},{"id":"dbda188d8d02fa9cf841f3fcdf55f962c1c327e7","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-06-03 11:11:06.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"206bb207244e4ddd95112b84dfaa26cb85ea698a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 13:00:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/38/661938/5/check/openstack-tox-docs/3e87464/html/ : SUCCESS in 5m 29s\n- openstack-tox-cover http://logs.openstack.org/38/661938/5/check/openstack-tox-cover/daddeef/cover/ : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661938/5/check/openstack-tox-lower-constraints/56cea8c/ : FAILURE in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/38/661938/5/check/openstack-tox-pep8/d545e93/ : SUCCESS in 5m 50s\n- openstack-tox-py27 http://logs.openstack.org/38/661938/5/check/openstack-tox-py27/89d0796/ : SUCCESS in 5m 45s\n- openstack-tox-py36 http://logs.openstack.org/38/661938/5/check/openstack-tox-py36/f99d1ad/ : SUCCESS in 6m 16s\n- openstack-tox-py37 http://logs.openstack.org/38/661938/5/check/openstack-tox-py37/200e5ff/ : SUCCESS in 5m 17s\n- tempest-full http://logs.openstack.org/38/661938/5/check/tempest-full/d6e43fd/ : SUCCESS in 1h 43m 27s\n- tempest-full-py3 http://logs.openstack.org/38/661938/5/check/tempest-full-py3/55b3c8a/ : SUCCESS in 1h 42m 33s","accounts_in_message":[],"_revision_number":5},{"id":"7c84853673f12090f576ef478b632c24f8c718ec","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-03 13:13:56.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1dbef9cad79ac17d25715c94a21ee601188ccdbd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 14:59:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/661938/5/check/openstack-tox-docs/cdc086a/html/ : SUCCESS in 6m 21s\n- openstack-tox-cover http://logs.openstack.org/38/661938/5/check/openstack-tox-cover/a20a913/cover/ : SUCCESS in 7m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661938/5/check/openstack-tox-lower-constraints/c917283/ : SUCCESS in 7m 08s\n- openstack-tox-pep8 http://logs.openstack.org/38/661938/5/check/openstack-tox-pep8/c40ad6e/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/38/661938/5/check/openstack-tox-py27/f60eb15/ : SUCCESS in 4m 45s\n- openstack-tox-py36 http://logs.openstack.org/38/661938/5/check/openstack-tox-py36/ec92795/ : SUCCESS in 5m 30s\n- openstack-tox-py37 http://logs.openstack.org/38/661938/5/check/openstack-tox-py37/62701c0/ : SUCCESS in 5m 08s\n- tempest-full http://logs.openstack.org/38/661938/5/check/tempest-full/792725b/ : SUCCESS in 1h 40m 19s\n- tempest-full-py3 http://logs.openstack.org/38/661938/5/check/tempest-full-py3/fc62c6e/ : SUCCESS in 1h 32m 00s","accounts_in_message":[],"_revision_number":5},{"id":"cb78212a84f654c56c4fdecfd2dca60c4a9a24d9","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-06-03 15:28:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"ee6e7661d516f432891c48fb79aac0d5ee3d0ea1","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-03 21:43:53.000000000","message":"Patch Set 5: Code-Review-1\n\nI am looking at the discussion in the Neutron patch that goes along with this one: https://review.opendev.org/#/c/661921. In the discussion over the Neutron side, it is not clear yet that we need this.\n\nAlso, boden requested in PS1 a release note and that comment has not been addressed","accounts_in_message":[],"_revision_number":5},{"id":"92886f8ccb1cdcafd26d2820496935ccad4bc98b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-04 06:21:20.000000000","message":"Removed Code-Review+2 by Slawek Kaplonski \u003cskaplons@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":5},{"id":"02b118828df78771a3a3119baa3a30fd91f1c2c7","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-04 12:20:58.000000000","message":"Patch Set 5: Code-Review-1\n\nPlease includes a release note.","accounts_in_message":[],"_revision_number":5},{"id":"857c8fa704dd56a154f589a4448243949099b2e8","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-06-05 19:21:40.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"8639a18c1d8d1b51ee6bad5bf27f79a2dfb0cc8a","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-06-12 01:41:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6c04aa0b8338b9730a371171c03a431f97a75f3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 03:55:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/661938/6/check/openstack-tox-docs/e2c05d9/html/ : SUCCESS in 5m 45s\n- openstack-tox-cover http://logs.openstack.org/38/661938/6/check/openstack-tox-cover/6aa57b6/cover/ : SUCCESS in 7m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661938/6/check/openstack-tox-lower-constraints/e19af17/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/38/661938/6/check/openstack-tox-pep8/c12548b/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/38/661938/6/check/openstack-tox-py27/b906dd9/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/38/661938/6/check/openstack-tox-py36/457649d/ : SUCCESS in 4m 49s\n- openstack-tox-py37 http://logs.openstack.org/38/661938/6/check/openstack-tox-py37/498f911/ : SUCCESS in 6m 23s\n- tempest-full http://logs.openstack.org/38/661938/6/check/tempest-full/ec895b0/ : SUCCESS in 1h 43m 43s\n- tempest-full-py3 http://logs.openstack.org/38/661938/6/check/tempest-full-py3/3517399/ : SUCCESS in 1h 34m 32s\n- build-openstack-releasenotes http://logs.openstack.org/38/661938/6/check/build-openstack-releasenotes/ad11a12/html/ : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":6},{"id":"06f769962a762259d4c18e718d59052795c78723","author":{"_account_id":18955,"name":"dharmendra","email":"dharmendra.kushwaha@gmail.com","username":"dharmendra"},"date":"2019-06-12 11:07:55.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"39483f89a99745d3025917d2a679276ffc9e87ef","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-12 12:31:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"89d15ba8341db9ace44cb2666eccf4ea7f23a663","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-13 14:52:08.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"077b9faa564e267049f600a978d19b08749279f9","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-06-14 05:13:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a77f8d10b65bcb147592a1ec476956a709a113d9","author":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"date":"2019-06-14 05:14:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bf97a07930a02590c476abbe654406f76b35ddec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 07:08:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/661938/7/check/openstack-tox-docs/aefb16f/html/ : SUCCESS in 4m 36s\n- openstack-tox-cover http://logs.openstack.org/38/661938/7/check/openstack-tox-cover/8f57c70/cover/ : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661938/7/check/openstack-tox-lower-constraints/bc18ca9/ : SUCCESS in 5m 39s\n- openstack-tox-pep8 http://logs.openstack.org/38/661938/7/check/openstack-tox-pep8/c729686/ : SUCCESS in 4m 53s\n- openstack-tox-py27 http://logs.openstack.org/38/661938/7/check/openstack-tox-py27/fe3a9f9/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/38/661938/7/check/openstack-tox-py36/bf132de/ : SUCCESS in 4m 44s\n- openstack-tox-py37 http://logs.openstack.org/38/661938/7/check/openstack-tox-py37/f07dcf3/ : SUCCESS in 12m 32s\n- tempest-full http://logs.openstack.org/38/661938/7/check/tempest-full/bc1014d/ : SUCCESS in 1h 34m 27s\n- tempest-full-py3 http://logs.openstack.org/38/661938/7/check/tempest-full-py3/a8c78dd/ : SUCCESS in 1h 34m 23s\n- build-openstack-releasenotes http://logs.openstack.org/38/661938/7/check/build-openstack-releasenotes/6323ce0/html/ : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":7},{"id":"e60691cebd232cb1580dd59e8610dff83232af13","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-14 09:32:14.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"0beb0e8cf727abfce34590ad9ddb20cd38038e36","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-06-14 11:45:55.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"28a189aed1dfce4f251c05bbe01a98588f6f2655","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-06-14 14:38:10.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"41a73b8e2112c4889db1de152e56f166002960b5","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-06-14 14:57:41.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"30e1029f91905291fc1a1371b6a06f83ff88e5a5","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-14 16:32:37.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"683dda3a7eff101fd2dc33767088564f1270831f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 16:32:50.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"eacedf9e6198ac9ee646b6ab0cd5f673184efc71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 18:18:22.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/38/661938/7/gate/openstack-tox-docs/d5d7837/html/ : SUCCESS in 4m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/38/661938/7/gate/openstack-tox-lower-constraints/0ea5822/ : SUCCESS in 5m 55s\n- openstack-tox-pep8 http://logs.openstack.org/38/661938/7/gate/openstack-tox-pep8/cd91c11/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/38/661938/7/gate/openstack-tox-py27/bdf61ad/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/38/661938/7/gate/openstack-tox-py36/bcd3704/ : SUCCESS in 7m 57s\n- openstack-tox-py37 http://logs.openstack.org/38/661938/7/gate/openstack-tox-py37/d256f05/ : SUCCESS in 7m 07s\n- tempest-full http://logs.openstack.org/38/661938/7/gate/tempest-full/5cb3dc7/ : SUCCESS in 1h 40m 07s\n- tempest-full-py3 http://logs.openstack.org/38/661938/7/gate/tempest-full-py3/41e69a0/ : SUCCESS in 1h 31m 30s\n- build-openstack-releasenotes http://logs.openstack.org/38/661938/7/gate/build-openstack-releasenotes/c3727c7/html/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":7},{"id":"15dfd1caf077481a068e1aff731930c7924649fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 18:18:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7d11cd6211a047d73c2a4795e8daa1b6a4851fa6","revisions":{"2e6191522b5790ea4494e273c51a9eff42b18576":{"kind":"REWORK","_number":1,"created":"2019-05-29 10:50:45.000000000","uploader":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"ref":"refs/changes/38/661938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/661938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/1"}}},"commit":{"parents":[{"commit":"96b82602fb4885747989b0914e823773dca5691e","subject":"sync sql fixtures from neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96b82602fb4885747989b0914e823773dca5691e"}]}],"author":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-29 09:21:53.000000000","tz":480},"committer":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-29 10:50:39.000000000","tz":480},"subject":"Add rarp protocol for neutron constants","message":"Add rarp protocol for neutron constants\n\nWe should support rarp protocol(0x8035) for live-migration\n\nChange-Id: I52c115acdb1a8c464f5fc6273f6d8c4d0c6526e9\nCloses-Bug: #1815989\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2e6191522b5790ea4494e273c51a9eff42b18576"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2e6191522b5790ea4494e273c51a9eff42b18576"}]},"branch":"refs/heads/master"},"14e1f734fe01f24490b7d7f0cf6f55c0ff372ee5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-31 02:41:31.000000000","uploader":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"ref":"refs/changes/38/661938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/661938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/2"}}},"commit":{"parents":[{"commit":"96b82602fb4885747989b0914e823773dca5691e","subject":"sync sql fixtures from neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96b82602fb4885747989b0914e823773dca5691e"}]}],"author":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-29 09:21:53.000000000","tz":480},"committer":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-31 02:40:41.000000000","tz":480},"subject":"Add rarp protocol for neutron constants","message":"Add rarp protocol for neutron constants\n\nWe should support rarp protocol(0x8035) for live-migration\n\nChange-Id: I52c115acdb1a8c464f5fc6273f6d8c4d0c6526e9\nPartial-Bug: #1815989\nSee-Also: https://review.openstack.org/661921 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/14e1f734fe01f24490b7d7f0cf6f55c0ff372ee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/14e1f734fe01f24490b7d7f0cf6f55c0ff372ee5"}]},"branch":"refs/heads/master"},"636b4351b3420a54a45038a9517279fc08d9c48d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-31 11:26:44.000000000","uploader":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"ref":"refs/changes/38/661938/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/661938/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/3"}}},"commit":{"parents":[{"commit":"96b82602fb4885747989b0914e823773dca5691e","subject":"sync sql fixtures from neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96b82602fb4885747989b0914e823773dca5691e"}]}],"author":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-29 09:21:53.000000000","tz":480},"committer":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-31 09:41:17.000000000","tz":0},"subject":"Add rarp protocol for neutron constants","message":"Add rarp protocol for neutron constants\n\nWe should support rarp protocol(0x8035) for live-migration\n\nChange-Id: I52c115acdb1a8c464f5fc6273f6d8c4d0c6526e9\nPartial-Bug: #1414559\nSee-Also: https://review.openstack.org/661921 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/636b4351b3420a54a45038a9517279fc08d9c48d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/636b4351b3420a54a45038a9517279fc08d9c48d"}]},"branch":"refs/heads/master"},"fe1b1f9253aac63dd9317800a01a7f7c71a3f054":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-06-03 02:35:21.000000000","uploader":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"ref":"refs/changes/38/661938/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/661938/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/4"}}},"commit":{"parents":[{"commit":"96b82602fb4885747989b0914e823773dca5691e","subject":"sync sql fixtures from neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96b82602fb4885747989b0914e823773dca5691e"}]}],"author":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-29 09:21:53.000000000","tz":480},"committer":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-06-03 02:35:17.000000000","tz":0},"subject":"Add rarp protocol for neutron constants","message":"Add rarp protocol for neutron constants\n\nWe should support rarp protocol(0x8035) for live-migration\n\nChange-Id: I52c115acdb1a8c464f5fc6273f6d8c4d0c6526e9\nPartial-Bug: #1831404\nSee-Also: https://review.openstack.org/661921 (neutron spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe1b1f9253aac63dd9317800a01a7f7c71a3f054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/fe1b1f9253aac63dd9317800a01a7f7c71a3f054"}]},"branch":"refs/heads/master"},"0beb014c018c1504cde1f82627fdbc0efd6dd54e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-06-03 11:11:06.000000000","uploader":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"ref":"refs/changes/38/661938/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/661938/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/5"}}},"commit":{"parents":[{"commit":"96b82602fb4885747989b0914e823773dca5691e","subject":"sync sql fixtures from neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96b82602fb4885747989b0914e823773dca5691e"}]}],"author":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-29 09:21:53.000000000","tz":480},"committer":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-06-03 11:10:29.000000000","tz":480},"subject":"Add rarp protocol for neutron constants","message":"Add rarp protocol for neutron constants\n\nWe should support rarp protocol(0x8035) for live-migration\n\nChange-Id: I52c115acdb1a8c464f5fc6273f6d8c4d0c6526e9\nPartial-Bug: #1831404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0beb014c018c1504cde1f82627fdbc0efd6dd54e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0beb014c018c1504cde1f82627fdbc0efd6dd54e"}]},"branch":"refs/heads/master"},"2042f3af37983f1d4e7e393edd7c1fca85f2d107":{"kind":"REWORK","_number":6,"created":"2019-06-12 01:41:29.000000000","uploader":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"ref":"refs/changes/38/661938/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/661938/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/6"}}},"commit":{"parents":[{"commit":"96b82602fb4885747989b0914e823773dca5691e","subject":"sync sql fixtures from neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96b82602fb4885747989b0914e823773dca5691e"}]}],"author":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-29 09:21:53.000000000","tz":480},"committer":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-06-12 01:40:56.000000000","tz":480},"subject":"Add rarp protocol for neutron constants","message":"Add rarp protocol for neutron constants\n\nWe should support rarp protocol(0x8035) for live-migration\nA release note is also included.\n\nChange-Id: I52c115acdb1a8c464f5fc6273f6d8c4d0c6526e9\nPartial-Bug: #1831404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2042f3af37983f1d4e7e393edd7c1fca85f2d107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/2042f3af37983f1d4e7e393edd7c1fca85f2d107"}]},"branch":"refs/heads/master"},"7d11cd6211a047d73c2a4795e8daa1b6a4851fa6":{"kind":"REWORK","_number":7,"created":"2019-06-14 05:14:34.000000000","uploader":{"_account_id":19956,"name":"Yang Li","email":"yang.li@easystack.cn","username":"leonstack"},"ref":"refs/changes/38/661938/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/38/661938/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/38/661938/7"}}},"commit":{"parents":[{"commit":"96b82602fb4885747989b0914e823773dca5691e","subject":"sync sql fixtures from neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/96b82602fb4885747989b0914e823773dca5691e"}]}],"author":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-05-29 09:21:53.000000000","tz":480},"committer":{"name":"Yang Li","email":"yang.li@easystack.cn","date":"2019-06-14 05:14:12.000000000","tz":480},"subject":"Add rarp protocol for neutron constants","message":"Add rarp protocol for neutron constants\n\nWe should support rarp protocol(0x8035) for live-migration\nA release note is also included.\n\nChange-Id: I52c115acdb1a8c464f5fc6273f6d8c4d0c6526e9\nPartial-Bug: #1831404\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7d11cd6211a047d73c2a4795e8daa1b6a4851fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7d11cd6211a047d73c2a4795e8daa1b6a4851fa6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
