)]}'
{"id":"openstack%2Fneutron-lib~665658","triplet_id":"openstack%2Fneutron-lib~master~Id10b06bd980a3e4c2accd14388bfcfabe37033d8","project":"openstack/neutron-lib","branch":"master","topic":"fix-subnet-ext","hashtags":[],"change_id":"Id10b06bd980a3e4c2accd14388bfcfabe37033d8","subject":"Fix subnet_dns_publish_fixed_ip extension","status":"MERGED","created":"2019-06-17 12:45:49.000000000","updated":"2019-06-18 14:06:48.000000000","submitted":"2019-06-18 14:06:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665658-1560866808333-44d5107d","meta_rev_id":"098ac9bae0007a3e731b8320a482efa166cffb49","_number":665658,"virtual_id_number":665658,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-17 13:42:30.000000000","_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-06-18 14:06:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2019-06-17 22:06:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-06-17 20:59:17.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2019-06-18 12:08:34.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 13:42:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-06-17 22:06:55.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-06-18 12:08:34.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-06-18 14:06:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c122a8c71ee1f9bf038e47c24ff646aae2c5f4fc","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-06-17 12:45:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c30c4fe832130a182afdb5ac27eaab324fe05749","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-06-17 13:42:30.000000000","message":"Patch Set 1:\n\nCan you provide a bug in the description? According to IRC comments, the problem if [1] is because the n-lib needed is not released yet.\n\n[1] http://logs.openstack.org/09/662409/9/check/neutron-tempest-dvr/d667254/controller/logs/screen-q-svc.txt.gz#_Jun_17_11_06_27_597034","accounts_in_message":[],"_revision_number":1},{"id":"7bcb225b6d19929a6b66e76c4dcc0c3f63d7add7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 14:52:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/665658/1/check/openstack-tox-docs/d276465/html/ : SUCCESS in 4m 31s\n- openstack-tox-cover http://logs.openstack.org/58/665658/1/check/openstack-tox-cover/ff2c698/cover/ : SUCCESS in 5m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/665658/1/check/openstack-tox-lower-constraints/862e1b9/ : SUCCESS in 5m 58s\n- openstack-tox-pep8 http://logs.openstack.org/58/665658/1/check/openstack-tox-pep8/9eca623/ : SUCCESS in 5m 58s\n- openstack-tox-py27 http://logs.openstack.org/58/665658/1/check/openstack-tox-py27/134b85d/ : SUCCESS in 5m 27s\n- openstack-tox-py36 http://logs.openstack.org/58/665658/1/check/openstack-tox-py36/e8e0116/ : SUCCESS in 4m 43s\n- openstack-tox-py37 http://logs.openstack.org/58/665658/1/check/openstack-tox-py37/8e5d522/ : SUCCESS in 5m 02s\n- tempest-full http://logs.openstack.org/58/665658/1/check/tempest-full/2b09f43/ : SUCCESS in 1h 43m 34s\n- tempest-full-py3 http://logs.openstack.org/58/665658/1/check/tempest-full-py3/fe0dba2/ : SUCCESS in 1h 35m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5623b7b90e716f9b2843ab7880389cd5f2d6e7e4","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-17 20:59:17.000000000","message":"Patch Set 1: Code-Review+2\n\nI think it maybe the case that since the subnet ext is part of the \"base\" or \"core\" resources that it\u0027s assumed to be there and not publicized by implementing plugins. I mean, not even the subnet_service_types api def marks subnet as required [1].\n\n\nI\u0027m fine with removing it as done here-in.\n\n\n[1] https://opendev.org/openstack/neutron-lib/src/branch/master/neutron_lib/api/definitions/subnet_service_types.py#L42","accounts_in_message":[],"_revision_number":1},{"id":"0ba1abccd496c798c423052f754c371135b1535a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-17 22:06:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4c4c3e3cd8983b2be9775a29f24809060695f33e","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-18 12:08:34.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f195cae3a1024487a778f607374f830d4d2ef18d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 12:08:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc6e4e287505e6f9f4c918f2b49dc83e9c743d5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:06:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/58/665658/1/gate/openstack-tox-docs/6549e27/html/ : SUCCESS in 7m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/58/665658/1/gate/openstack-tox-lower-constraints/f30e885/ : SUCCESS in 5m 22s\n- openstack-tox-pep8 http://logs.openstack.org/58/665658/1/gate/openstack-tox-pep8/65df3de/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/58/665658/1/gate/openstack-tox-py27/817735a/ : SUCCESS in 4m 24s\n- openstack-tox-py36 http://logs.openstack.org/58/665658/1/gate/openstack-tox-py36/e486f3e/ : SUCCESS in 4m 30s\n- openstack-tox-py37 http://logs.openstack.org/58/665658/1/gate/openstack-tox-py37/931a394/ : SUCCESS in 5m 34s\n- tempest-full http://logs.openstack.org/58/665658/1/gate/tempest-full/20b45a1/ : SUCCESS in 1h 54m 42s\n- tempest-full-py3 http://logs.openstack.org/58/665658/1/gate/tempest-full-py3/82f8433/ : SUCCESS in 1h 32m 08s","accounts_in_message":[],"_revision_number":1},{"id":"57f3138713a74231fc4bf852828b907a9f0b791b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 14:06:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b","revisions":{"d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b":{"kind":"REWORK","_number":1,"created":"2019-06-17 12:45:49.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/58/665658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/58/665658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/58/665658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/58/665658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/58/665658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/58/665658/1"}}},"commit":{"parents":[{"commit":"893f030b0919123322b56ca440323cb97c5b3d81","subject":"Merge \"Add rarp protocol for neutron constants\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/893f030b0919123322b56ca440323cb97c5b3d81"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-06-17 12:44:46.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-06-17 12:44:46.000000000","tz":0},"subject":"Fix subnet_dns_publish_fixed_ip extension","message":"Fix subnet_dns_publish_fixed_ip extension\n\nDepending on ``subnet`` as required extension doesn\u0027t work because it\ndoesn\u0027t exist.\n\nChange-Id: Id10b06bd980a3e4c2accd14388bfcfabe37033d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
