)]}'
{"id":"openstack%2Fneutron-lib~667070","triplet_id":"openstack%2Fneutron-lib~master~Ifa545242224bdc80a934b529e44b25b4492d4e0b","project":"openstack/neutron-lib","branch":"master","topic":"bug/1834012","hashtags":[],"change_id":"Ifa545242224bdc80a934b529e44b25b4492d4e0b","subject":"Loopback address routing should be invalid","status":"MERGED","created":"2019-06-24 11:53:51.000000000","updated":"2019-07-01 08:49:31.000000000","submitted":"2019-06-26 22:59:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667070-1561589953887-f790302f","meta_rev_id":"8c1d4fea96ed360d30e01dd041fba05a73d6cd6d","_number":667070,"virtual_id_number":667070,"owner":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"date":"2019-06-26 14:35:39.000000000","_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-06-26 22:59:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-25 09:04:46.000000000","_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"recommended":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"all":[{"value":1,"date":"2019-06-25 09:24:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-06-25 12:15:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":2,"date":"2019-06-26 21:03:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-06-26 15:19:02.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":1,"date":"2019-06-26 21:03:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-25 09:24:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-06-25 12:15:23.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2019-06-26 14:35:39.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-06-26 15:19:02.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-06-26 21:03:55.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-06-26 22:59:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c2455b0d4da4f2aeac3f5e92a78b6b8dd6436e7","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2019-06-24 11:53:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e6ab886cda2e5e835df9d9844f27b39609780858","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2019-06-24 11:56:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7105174b501f49a75ff4f9d121889cc9162b3cf7","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-24 12:25:12.000000000","message":"Patch Set 2: Code-Review-1\n\nCan you please also add a release note to this. Since it\u0027s fixing a bug, probably best to add the note under the \u0027bugs\u0027 section of the reno.","accounts_in_message":[],"_revision_number":2},{"id":"165a2d39b5b2a384576a4d754c9f0de94a09f54a","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-06-24 12:27:43.000000000","message":"Patch Set 2: Code-Review-1\n\nBeyond the release note please also add a unit test to neutron_lib/tests/unit/api/validators/test_validators.py.","accounts_in_message":[],"_revision_number":2},{"id":"aec7e12cacedc993f3cce89495d86cfe8d4125ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 13:38:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/70/667070/2/check/openstack-tox-docs/8a839ab/html/ : SUCCESS in 7m 09s\n- openstack-tox-cover http://logs.openstack.org/70/667070/2/check/openstack-tox-cover/1f88de4/cover/ : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/667070/2/check/openstack-tox-lower-constraints/9aa32e8/ : SUCCESS in 5m 05s\n- openstack-tox-pep8 http://logs.openstack.org/70/667070/2/check/openstack-tox-pep8/53d4e4f/ : SUCCESS in 5m 00s\n- openstack-tox-py27 http://logs.openstack.org/70/667070/2/check/openstack-tox-py27/3f09dbe/ : SUCCESS in 4m 53s\n- openstack-tox-py36 http://logs.openstack.org/70/667070/2/check/openstack-tox-py36/bc2b2b6/ : SUCCESS in 4m 54s\n- openstack-tox-py37 http://logs.openstack.org/70/667070/2/check/openstack-tox-py37/1c61ec3/ : SUCCESS in 5m 10s\n- tempest-full http://logs.openstack.org/70/667070/2/check/tempest-full/de563cf/ : FAILURE in 1h 36m 26s\n- tempest-full-py3 http://logs.openstack.org/70/667070/2/check/tempest-full-py3/e151249/ : SUCCESS in 1h 32m 44s","accounts_in_message":[],"_revision_number":2},{"id":"975d819206243e92ec419b82c90cd063387f886e","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2019-06-25 07:25:50.000000000","message":"Patch Set 2:\n\nrecehck","accounts_in_message":[],"_revision_number":2},{"id":"9bf74d5a3010fbe9160faef466341fca12f805cc","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2019-06-25 09:04:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7bd5126642958046cee5338f3d7593e0f4b523a4","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2019-06-25 09:04:46.000000000","message":"Patch Set 3:\n\nDone","accounts_in_message":[],"_revision_number":3},{"id":"89ea022b9f283419d56ba02e22e5eeefb3247505","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-06-25 09:24:42.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"fd3a432b451354dece0a360c5a8a730927a9ed56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 12:14:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/70/667070/3/check/openstack-tox-docs/b7e0128/html/ : SUCCESS in 8m 39s\n- openstack-tox-cover http://logs.openstack.org/70/667070/3/check/openstack-tox-cover/9da483b/cover/ : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/667070/3/check/openstack-tox-lower-constraints/0c2e174/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/70/667070/3/check/openstack-tox-pep8/31e085e/ : SUCCESS in 5m 48s\n- openstack-tox-py27 http://logs.openstack.org/70/667070/3/check/openstack-tox-py27/1e1fbd5/ : SUCCESS in 6m 22s\n- openstack-tox-py36 http://logs.openstack.org/70/667070/3/check/openstack-tox-py36/bce84d3/ : SUCCESS in 5m 37s\n- openstack-tox-py37 http://logs.openstack.org/70/667070/3/check/openstack-tox-py37/590fc2a/ : SUCCESS in 6m 35s\n- tempest-full finger://ze07.openstack.org/d9e921f19e994037aae6061f6c1b5e88 : RETRY_LIMIT in 57m 57s\n- tempest-full-py3 finger://ze07.openstack.org/3f393e1638304decbd0e670edc73e7bb : RETRY_LIMIT in 31m 53s\n- build-openstack-releasenotes http://logs.openstack.org/70/667070/3/check/build-openstack-releasenotes/89345e3/html/ : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":3},{"id":"a283e88203dec8894f91aeccb4bddc79e117c419","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-06-25 12:15:23.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"460a8fc7388cb16b4caef18540709d748d3f086b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 15:04:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/70/667070/3/check/openstack-tox-docs/5a97ff6/html/ : SUCCESS in 2m 50s\n- openstack-tox-cover http://logs.openstack.org/70/667070/3/check/openstack-tox-cover/3496e60/cover/ : SUCCESS in 6m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/667070/3/check/openstack-tox-lower-constraints/c87a95c/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/70/667070/3/check/openstack-tox-pep8/f1a1939/ : SUCCESS in 6m 01s\n- openstack-tox-py27 http://logs.openstack.org/70/667070/3/check/openstack-tox-py27/624d7d3/ : SUCCESS in 4m 03s\n- openstack-tox-py36 http://logs.openstack.org/70/667070/3/check/openstack-tox-py36/ee3fecf/ : SUCCESS in 5m 19s\n- openstack-tox-py37 http://logs.openstack.org/70/667070/3/check/openstack-tox-py37/8a5e217/ : SUCCESS in 3m 48s\n- tempest-full finger://ze08.openstack.org/bb03e3cb7e9646819243fd08543d4e51 : RETRY_LIMIT in 24m 02s\n- tempest-full-py3 http://logs.openstack.org/70/667070/3/check/tempest-full-py3/944878b/ : RETRY_LIMIT in 21m 59s\n- build-openstack-releasenotes http://logs.openstack.org/70/667070/3/check/build-openstack-releasenotes/a744c09/html/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":3},{"id":"ce7526c82d0a68882a683b32b7da93c92798b766","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-26 12:10:41.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"12c94075406170e46e9bbc44357037ac20db38ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 13:57:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/667070/3/check/openstack-tox-docs/0312053/html/ : SUCCESS in 2m 36s\n- openstack-tox-cover http://logs.openstack.org/70/667070/3/check/openstack-tox-cover/6058a24/cover/ : SUCCESS in 3m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/667070/3/check/openstack-tox-lower-constraints/a81bb78/ : SUCCESS in 6m 24s\n- openstack-tox-pep8 http://logs.openstack.org/70/667070/3/check/openstack-tox-pep8/679443c/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/70/667070/3/check/openstack-tox-py27/ab4a8fd/ : SUCCESS in 4m 13s\n- openstack-tox-py36 http://logs.openstack.org/70/667070/3/check/openstack-tox-py36/8a5e700/ : SUCCESS in 3m 44s\n- openstack-tox-py37 http://logs.openstack.org/70/667070/3/check/openstack-tox-py37/a06db18/ : SUCCESS in 3m 54s\n- tempest-full http://logs.openstack.org/70/667070/3/check/tempest-full/71228e8/ : SUCCESS in 1h 43m 34s\n- tempest-full-py3 http://logs.openstack.org/70/667070/3/check/tempest-full-py3/a224562/ : SUCCESS in 1h 41m 08s\n- build-openstack-releasenotes http://logs.openstack.org/70/667070/3/check/build-openstack-releasenotes/1d7aa48/html/ : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":3},{"id":"d94f6c82ac66fd7527d907cbe662366305acda02","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-26 14:35:39.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nJust a nit on the release note.","accounts_in_message":[],"_revision_number":3},{"id":"b620eafea58c055c0d730e3e6125e33fdf6e2125","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-26 15:19:02.000000000","message":"Patch Set 3: Code-Review+2\n\nWhile it might be nice to fix the wording in the release note, I don\u0027t see a need to hold up the patch for that since we mainly just use those notes for tracking purposes and a ref to the bug is sufficient.","accounts_in_message":[],"_revision_number":3},{"id":"5c9038466c350346cb6e3b665208a6261b016e8a","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-06-26 21:03:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c07a564cb58b16169a4734fb7742dc279695fa7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 21:04:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5d953782e00daf2788bb95f4c86a81f3498b0fed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:59:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"58543b7ebf633aba6e40f4679b8eaa5420eb90ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:59:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/70/667070/3/gate/openstack-tox-docs/9c23e01/html/ : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/667070/3/gate/openstack-tox-lower-constraints/3f7763f/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/70/667070/3/gate/openstack-tox-pep8/e31f85d/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/70/667070/3/gate/openstack-tox-py27/e4d69bd/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/70/667070/3/gate/openstack-tox-py36/477fe15/ : SUCCESS in 3m 35s\n- openstack-tox-py37 http://logs.openstack.org/70/667070/3/gate/openstack-tox-py37/5c70bc3/ : SUCCESS in 3m 52s\n- tempest-full http://logs.openstack.org/70/667070/3/gate/tempest-full/57cda06/ : SUCCESS in 1h 50m 23s\n- tempest-full-py3 http://logs.openstack.org/70/667070/3/gate/tempest-full-py3/54f5850/ : SUCCESS in 1h 29m 32s\n- build-openstack-releasenotes http://logs.openstack.org/70/667070/3/gate/build-openstack-releasenotes/1ff14f1/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":3},{"id":"78a7cd75320453c17de8ef9a90bdc84a455d50a6","author":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"date":"2019-07-01 08:49:31.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 1648df46f6961891d5c78b6b24488f258b793e64","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8417717411894c2a2b315e21495245a83c23dc87","revisions":{"5362e031b85e41985b02e4d93ea157cf87edccad":{"kind":"REWORK","_number":1,"created":"2019-06-24 11:53:51.000000000","uploader":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"ref":"refs/changes/70/667070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/70/667070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/1"}}},"commit":{"parents":[{"commit":"d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b","subject":"Fix subnet_dns_publish_fixed_ip extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b"}]}],"author":{"name":"Kobi Samoray","email":"ksamoray@vmware.com","date":"2019-06-24 11:52:29.000000000","tz":180},"committer":{"name":"Kobi Samoray","email":"ksamoray@vmware.com","date":"2019-06-24 11:53:47.000000000","tz":180},"subject":"Loopback address routing should be invalid","message":"Loopback address routing should be invalid\n\nHost route validators should reject loopback CIDRs.\n\nChange-Id: Ifa545242224bdc80a934b529e44b25b4492d4e0b\nFixes-bug: 1834012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5362e031b85e41985b02e4d93ea157cf87edccad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/5362e031b85e41985b02e4d93ea157cf87edccad"}]},"branch":"refs/heads/master"},"7d4a3bb4deb11f8e92efc4b4916f38bf9dfcef08":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-24 11:56:01.000000000","uploader":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"ref":"refs/changes/70/667070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/70/667070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/2"}}},"commit":{"parents":[{"commit":"d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b","subject":"Fix subnet_dns_publish_fixed_ip extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b"}]}],"author":{"name":"Kobi Samoray","email":"ksamoray@vmware.com","date":"2019-06-24 11:52:29.000000000","tz":180},"committer":{"name":"Kobi Samoray","email":"ksamoray@vmware.com","date":"2019-06-24 11:55:46.000000000","tz":180},"subject":"Loopback address routing should be invalid","message":"Loopback address routing should be invalid\n\nHost route validators should reject loopback CIDRs.\n\nChange-Id: Ifa545242224bdc80a934b529e44b25b4492d4e0b\nCLoses-bug: 1834012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7d4a3bb4deb11f8e92efc4b4916f38bf9dfcef08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7d4a3bb4deb11f8e92efc4b4916f38bf9dfcef08"}]},"branch":"refs/heads/master"},"8417717411894c2a2b315e21495245a83c23dc87":{"kind":"REWORK","_number":3,"created":"2019-06-25 09:04:21.000000000","uploader":{"_account_id":13438,"name":"Kobi Samoray","email":"kobi.samoray@broadcom.com","username":"ksamoray"},"ref":"refs/changes/70/667070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/70/667070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/70/667070/3"}}},"commit":{"parents":[{"commit":"d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b","subject":"Fix subnet_dns_publish_fixed_ip extension","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d2cabd9ed8cd64f1840f3d003fbf6922eb094d4b"}]}],"author":{"name":"Kobi Samoray","email":"ksamoray@vmware.com","date":"2019-06-24 11:52:29.000000000","tz":180},"committer":{"name":"Kobi Samoray","email":"ksamoray@vmware.com","date":"2019-06-25 09:04:03.000000000","tz":180},"subject":"Loopback address routing should be invalid","message":"Loopback address routing should be invalid\n\nHost route validators should reject loopback CIDRs.\n\nChange-Id: Ifa545242224bdc80a934b529e44b25b4492d4e0b\nCLoses-bug: 1834012\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8417717411894c2a2b315e21495245a83c23dc87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/8417717411894c2a2b315e21495245a83c23dc87"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
