)]}'
{"id":"openstack%2Fneutron-lib~667644","triplet_id":"openstack%2Fneutron-lib~master~Id541059b8fe0ab74035819516f079d93bb76d43c","project":"openstack/neutron-lib","branch":"master","topic":"bug/1830240","hashtags":[],"change_id":"Id541059b8fe0ab74035819516f079d93bb76d43c","subject":"Introduce NetworkAddressScopeAffinityError","status":"MERGED","created":"2019-06-26 15:06:18.000000000","updated":"2019-06-27 13:57:08.000000000","submitted":"2019-06-27 13:57:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667644-1561643828428-3286e172","meta_rev_id":"88cb10e73fd0981b494bf439ceefb05edadbb3ac","_number":667644,"virtual_id_number":667644,"owner":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-06-27 13:57:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2019-06-27 11:56:32.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-06-27 08:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2019-06-27 11:56:32.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 15:45:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-06-26 17:29:44.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-06-27 08:38:20.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-06-27 11:56:32.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-06-27 13:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fb81fe5b243fc64454672164092bcc2550328dfb","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-26 15:06:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e833f3a71a9471c4bc172b0782d637de5d19e075","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-26 15:07:17.000000000","message":"Topic changed from bug/1830240 to bp/subnets-different-pools-same-net","accounts_in_message":[],"_revision_number":1},{"id":"4a13d9958fceb68673dec0330743597e0123b28e","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-26 15:20:13.000000000","message":"Patch Set 1: Code-Review-1\n\nCan you please add a quick release note with this?\nWe use those for tracking purposes.\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"2c22267f098e52e84186408bce045544c13770b1","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-26 15:29:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"49244307b3312189b13e978bbc585b900cd575ae","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-26 17:29:44.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a70357efb26252744a31b3c742596c79f9055c17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 17:33:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/667644/2/check/openstack-tox-docs/490eaca/html/ : SUCCESS in 2m 46s\n- openstack-tox-cover http://logs.openstack.org/44/667644/2/check/openstack-tox-cover/7001898/cover/ : SUCCESS in 4m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/44/667644/2/check/openstack-tox-lower-constraints/b2ab293/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 http://logs.openstack.org/44/667644/2/check/openstack-tox-pep8/8b7348d/ : SUCCESS in 4m 25s\n- openstack-tox-py27 http://logs.openstack.org/44/667644/2/check/openstack-tox-py27/5e082bf/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/44/667644/2/check/openstack-tox-py36/2179fe8/ : SUCCESS in 3m 31s\n- openstack-tox-py37 http://logs.openstack.org/44/667644/2/check/openstack-tox-py37/a2456cc/ : SUCCESS in 4m 11s\n- tempest-full http://logs.openstack.org/44/667644/2/check/tempest-full/3e43bc4/ : SUCCESS in 1h 57m 27s\n- tempest-full-py3 http://logs.openstack.org/44/667644/2/check/tempest-full-py3/7eadc6c/ : SUCCESS in 1h 53m 14s\n- build-openstack-releasenotes http://logs.openstack.org/44/667644/2/check/build-openstack-releasenotes/6f83210/html/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":2},{"id":"5b5ab3525a5baf864827aed9b065d94967b1f23a","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-26 18:35:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab4e2856bcf4abc25db862881c9a12c9e863270b","author":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"date":"2019-06-26 18:57:20.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"290632acc1ab1afaa71bdec2ee04959039f84f66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 21:07:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/667644/3/check/openstack-tox-docs/bae58b0/html/ : SUCCESS in 2m 54s\n- openstack-tox-cover http://logs.openstack.org/44/667644/3/check/openstack-tox-cover/87b9333/cover/ : SUCCESS in 3m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/44/667644/3/check/openstack-tox-lower-constraints/00513e1/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/44/667644/3/check/openstack-tox-pep8/a415430/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/44/667644/3/check/openstack-tox-py27/84d271a/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/44/667644/3/check/openstack-tox-py36/2ef0738/ : SUCCESS in 3m 20s\n- openstack-tox-py37 http://logs.openstack.org/44/667644/3/check/openstack-tox-py37/781fb4b/ : SUCCESS in 4m 04s\n- tempest-full http://logs.openstack.org/44/667644/3/check/tempest-full/e726ad7/ : SUCCESS in 2h 03m 02s\n- tempest-full-py3 http://logs.openstack.org/44/667644/3/check/tempest-full-py3/bc56e3d/ : SUCCESS in 1h 30m 32s\n- build-openstack-releasenotes http://logs.openstack.org/44/667644/3/check/build-openstack-releasenotes/ffb0d9e/html/ : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":3},{"id":"9b62cb8b92bb03025cecaaf8b0fc761c84786b8d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-27 08:38:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"159a26225c844a0fa852d0d8451cb9f257d50413","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-27 11:56:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"baf812fa51d420195724fc96f97785e5b592b783","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:56:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3f3fa38e3402f28380efdf23e02f3eacec9114f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:57:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/667644/3/gate/openstack-tox-docs/be3c8e1/html/ : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/44/667644/3/gate/openstack-tox-lower-constraints/35c2658/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/44/667644/3/gate/openstack-tox-pep8/adf4962/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/44/667644/3/gate/openstack-tox-py27/1f3bfea/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/44/667644/3/gate/openstack-tox-py36/149ab0f/ : SUCCESS in 3m 43s\n- openstack-tox-py37 http://logs.openstack.org/44/667644/3/gate/openstack-tox-py37/41959fc/ : SUCCESS in 3m 46s\n- tempest-full http://logs.openstack.org/44/667644/3/gate/tempest-full/00c8cdc/ : SUCCESS in 1h 52m 36s\n- tempest-full-py3 http://logs.openstack.org/44/667644/3/gate/tempest-full-py3/005c3bc/ : SUCCESS in 1h 53m 42s\n- build-openstack-releasenotes http://logs.openstack.org/44/667644/3/gate/build-openstack-releasenotes/4b35a68/html/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":3},{"id":"ba9d7cc2e702936c073644a9dec785e2e9016564","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 13:57:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d4415186690493c728e461d9bea9b32bfe246321","revisions":{"e9a33595713bfe953c3102374139741b3223ac9c":{"kind":"REWORK","_number":1,"created":"2019-06-26 15:06:18.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/44/667644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/44/667644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/1"}}},"commit":{"parents":[{"commit":"c26573f847958fc090ba55b9c086a1cc5756b15d","subject":"Add new vif_details parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c26573f847958fc090ba55b9c086a1cc5756b15d"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 15:04:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 15:04:00.000000000","tz":-300},"subject":"Introduce NetworkAddressScopeAffinityError","message":"Introduce NetworkAddressScopeAffinityError\n\nThis introduces NetworkAddressScopeAffinityError, which is to be\nraised when address scope affinity constraints on a network are\nviolated.\n\nChange-Id: Id541059b8fe0ab74035819516f079d93bb76d43c\nPartial-Bug: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9a33595713bfe953c3102374139741b3223ac9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e9a33595713bfe953c3102374139741b3223ac9c"}]},"branch":"refs/heads/master"},"0b62cfa4358047185f9a471355486de74ea25b4a":{"kind":"REWORK","_number":2,"created":"2019-06-26 15:29:24.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/44/667644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/44/667644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/2"}}},"commit":{"parents":[{"commit":"c26573f847958fc090ba55b9c086a1cc5756b15d","subject":"Add new vif_details parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c26573f847958fc090ba55b9c086a1cc5756b15d"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 15:04:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 15:29:12.000000000","tz":-300},"subject":"Introduce NetworkAddressScopeAffinityError","message":"Introduce NetworkAddressScopeAffinityError\n\nThis introduces NetworkAddressScopeAffinityError, which is to be\nraised when address scope affinity constraints on a network are\nviolated.\n\nChange-Id: Id541059b8fe0ab74035819516f079d93bb76d43c\nPartial-Bug: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0b62cfa4358047185f9a471355486de74ea25b4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/0b62cfa4358047185f9a471355486de74ea25b4a"}]},"branch":"refs/heads/master"},"d4415186690493c728e461d9bea9b32bfe246321":{"kind":"REWORK","_number":3,"created":"2019-06-26 18:57:20.000000000","uploader":{"_account_id":4187,"name":"Ryan Tidwell","email":"rtidwell@suse.com","username":"ryan-tidwell"},"ref":"refs/changes/44/667644/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/44/667644/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/44/667644/3"}}},"commit":{"parents":[{"commit":"c26573f847958fc090ba55b9c086a1cc5756b15d","subject":"Add new vif_details parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c26573f847958fc090ba55b9c086a1cc5756b15d"}]}],"author":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 15:04:00.000000000","tz":-300},"committer":{"name":"Ryan Tidwell","email":"rtidwell@suse.com","date":"2019-06-26 18:57:00.000000000","tz":0},"subject":"Introduce NetworkAddressScopeAffinityError","message":"Introduce NetworkAddressScopeAffinityError\n\nThis introduces NetworkAddressScopeAffinityError, which is to be\nraised when address scope affinity constraints on a network are\nviolated.\n\nChange-Id: Id541059b8fe0ab74035819516f079d93bb76d43c\nPartial-Bug: 1830240\nImplements: subnets-different-pools-same-net\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d4415186690493c728e461d9bea9b32bfe246321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d4415186690493c728e461d9bea9b32bfe246321"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
