)]}'
{"id":"openstack%2Fneutron-lib~667723","triplet_id":"openstack%2Fneutron-lib~master~Ibd9cff863eaff940755ccb9c20f39e29e2001cde","project":"openstack/neutron-lib","branch":"master","topic":"new-flake8","hashtags":[],"change_id":"Ibd9cff863eaff940755ccb9c20f39e29e2001cde","subject":"Update flake8 style enforcement","status":"MERGED","created":"2019-06-26 20:31:36.000000000","updated":"2019-06-28 08:25:08.000000000","submitted":"2019-06-27 14:00:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667723-1561644006910-90acf6b2","meta_rev_id":"69b76b6c94a926ffc3cecef46a7e35a66f87efb0","_number":667723,"virtual_id_number":667723,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2019-06-27 14:00:06.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2019-06-27 07:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-06-27 11:57:45.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":1,"date":"2019-06-27 06:38:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-06-27 11:57:45.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 06:38:10.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-06-27 07:17:26.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-06-27 11:57:45.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-06-27 14:00:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37dd14bcb5041c5313e9434b29015d6d8c88eb14","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-26 20:31:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f242bb76c4e153c9085f3da63d80c0876fc09a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 22:29:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/23/667723/1/check/build-openstack-api-ref/1fcfadd/html/ : SUCCESS in 3m 24s\n- requirements-check http://logs.openstack.org/23/667723/1/check/requirements-check/3f7e47f/ : SUCCESS in 1m 56s\n- openstack-tox-docs http://logs.openstack.org/23/667723/1/check/openstack-tox-docs/a5ae305/html/ : SUCCESS in 3m 25s\n- openstack-tox-cover http://logs.openstack.org/23/667723/1/check/openstack-tox-cover/ea10469/cover/ : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/667723/1/check/openstack-tox-lower-constraints/1ff4d67/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/23/667723/1/check/openstack-tox-pep8/2b53b20/ : SUCCESS in 4m 26s\n- openstack-tox-py27 http://logs.openstack.org/23/667723/1/check/openstack-tox-py27/ac75123/ : SUCCESS in 3m 26s\n- openstack-tox-py36 http://logs.openstack.org/23/667723/1/check/openstack-tox-py36/7ce72d8/ : SUCCESS in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/23/667723/1/check/openstack-tox-py37/40f427f/ : SUCCESS in 4m 35s\n- tempest-full http://logs.openstack.org/23/667723/1/check/tempest-full/4d4a843/ : SUCCESS in 1h 49m 52s\n- tempest-full-py3 http://logs.openstack.org/23/667723/1/check/tempest-full-py3/32ff085/ : SUCCESS in 1h 30m 40s\n- build-openstack-releasenotes http://logs.openstack.org/23/667723/1/check/build-openstack-releasenotes/de29d70/html/ : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9849fd729aedeaec1fc48caea7976c5c8bf86e5f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-06-27 06:38:10.000000000","message":"Patch Set 1: Code-Review+1\n\ndo you think that the same checks \u0026 checker version should be used on stadium projects as well to make py3 change smoother?","accounts_in_message":[],"_revision_number":1},{"id":"adc5f01f3a310ab859a12806a8bea3d4d7402818","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-06-27 07:17:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"36dcf2219ba68c18a21bc6ade08d8881251b7554","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-27 11:57:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"314332a67a8489873290e748f05d3b09b3f34074","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 11:57:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5900f87a818fa28cde6a68b598325f2ee8be10cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:00:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ce2ef6ce76c17a5a9774432389c1b846c734158d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 14:00:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/23/667723/1/gate/build-openstack-api-ref/ad2cd05/html/ : SUCCESS in 3m 24s\n- requirements-check http://logs.openstack.org/23/667723/1/gate/requirements-check/2367c73/ : SUCCESS in 2m 01s\n- openstack-tox-docs http://logs.openstack.org/23/667723/1/gate/openstack-tox-docs/4e6c721/html/ : SUCCESS in 3m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/667723/1/gate/openstack-tox-lower-constraints/1ceb0ed/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/23/667723/1/gate/openstack-tox-pep8/f091aba/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/23/667723/1/gate/openstack-tox-py27/efc43d7/ : SUCCESS in 4m 28s\n- openstack-tox-py36 http://logs.openstack.org/23/667723/1/gate/openstack-tox-py36/8f61b24/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/23/667723/1/gate/openstack-tox-py37/8bf4792/ : SUCCESS in 4m 02s\n- tempest-full http://logs.openstack.org/23/667723/1/gate/tempest-full/6233a3e/ : SUCCESS in 1h 55m 27s\n- tempest-full-py3 http://logs.openstack.org/23/667723/1/gate/tempest-full-py3/90538ee/ : SUCCESS in 1h 49m 28s\n- build-openstack-releasenotes http://logs.openstack.org/23/667723/1/gate/build-openstack-releasenotes/243c0b7/html/ : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"af41c28597a28f34b36b9270907bc29559697258","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-06-27 15:27:03.000000000","message":"Patch Set 1:\n\n\u003e do you think that the same checks \u0026 checker version should be used on stadium projects as well to make py3 change smoother?\n\nLajos - we should be trying to keep them close, otherwise we start getting broken by accident when the requirements repo moves things forward.  I\u0027m not sure where vpnaas and fwaas are (or others), I\u0027ve mostly been keeping neutron, neutron-tempest-plugin and networking-ovn up to date myself.","accounts_in_message":[],"_revision_number":1},{"id":"f84f7613f3e97b4ab06f472afa3ef357522fdcc1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-06-28 08:25:08.000000000","message":"Patch Set 1:\n\nBrian: thanks, I check networking-l2gw/odl","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf810d24ed715910cf6b10e4fe995315a8c3bb5a","revisions":{"bf810d24ed715910cf6b10e4fe995315a8c3bb5a":{"kind":"REWORK","_number":1,"created":"2019-06-26 20:31:36.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/23/667723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/23/667723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/667723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/667723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/23/667723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/23/667723/1"}}},"commit":{"parents":[{"commit":"c26573f847958fc090ba55b9c086a1cc5756b15d","subject":"Add new vif_details parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c26573f847958fc090ba55b9c086a1cc5756b15d"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-06-26 20:30:35.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-06-26 20:30:35.000000000","tz":-240},"subject":"Update flake8 style enforcement","message":"Update flake8 style enforcement\n\nAdded enforcement of H106, H203, H204 and H205 extensions to\nmatch neutron repo, fixing a few failures in the process.\n\nAlso updated to support hacking 1.1.0, bringing in flake8\n2.6.2 and pycodestyle 2.4.0 in the process.  This ensures\nfuture python3 compatibility and a bit better code styling.\n\nChange-Id: Ibd9cff863eaff940755ccb9c20f39e29e2001cde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bf810d24ed715910cf6b10e4fe995315a8c3bb5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bf810d24ed715910cf6b10e4fe995315a8c3bb5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
