)]}'
{"id":"openstack%2Fneutron-lib~668196","triplet_id":"openstack%2Fneutron-lib~master~I5e9497d650176201407f0823f0fce46e009fb89c","project":"openstack/neutron-lib","branch":"master","topic":"bp/neutronlib-decouple-db","hashtags":[],"change_id":"I5e9497d650176201407f0823f0fce46e009fb89c","subject":"make sql fixtures public","status":"MERGED","created":"2019-06-28 15:39:46.000000000","updated":"2019-07-01 22:09:30.000000000","submitted":"2019-07-01 22:09:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668196-1562018970147-16fc3d50","meta_rev_id":"8aa71b40c1babe3b1d52ec0f7ae43561e72829f1","_number":668196,"virtual_id_number":668196,"owner":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-07-01 22:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2019-07-01 19:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-07-01 08:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2019-07-01 19:59:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 08:39:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-07-01 19:59:56.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-07-01 22:09:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b1d63448d08cf724ff985c000c97ea74a7c857a","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-06-28 15:39:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77eb4d49624450581a8806c0c8b14bf64422e8f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 17:30:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/668196/1/check/openstack-tox-docs/0f92283/html/ : SUCCESS in 2m 21s\n- openstack-tox-cover http://logs.openstack.org/96/668196/1/check/openstack-tox-cover/b059110/cover/ : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/668196/1/check/openstack-tox-lower-constraints/dd5c69a/ : SUCCESS in 3m 43s\n- openstack-tox-pep8 http://logs.openstack.org/96/668196/1/check/openstack-tox-pep8/4c87f50/ : SUCCESS in 3m 15s\n- openstack-tox-py27 http://logs.openstack.org/96/668196/1/check/openstack-tox-py27/d84653b/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/96/668196/1/check/openstack-tox-py36/8476b21/ : SUCCESS in 2m 57s\n- openstack-tox-py37 http://logs.openstack.org/96/668196/1/check/openstack-tox-py37/b959197/ : SUCCESS in 3m 46s\n- tempest-full http://logs.openstack.org/96/668196/1/check/tempest-full/b5aadf0/ : SUCCESS in 1h 49m 01s\n- tempest-full-py3 http://logs.openstack.org/96/668196/1/check/tempest-full-py3/6b231a5/ : SUCCESS in 1h 32m 30s\n- build-openstack-releasenotes http://logs.openstack.org/96/668196/1/check/build-openstack-releasenotes/21973c0/html/ : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3b337bdb8637ef5009196a283df62ce7f5eaca38","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-01 08:39:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"24812e5c4fb06e29dcc53466de8c6d6d1d21c13b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-07-01 19:59:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"640bdb2f56794d1174dec8ffecc1377833e4d6b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 20:00:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21ea99c311310fff3a931ef6441c0bda18eefde8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 22:09:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/96/668196/1/gate/openstack-tox-docs/0f3e9cb/html/ : SUCCESS in 4m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/668196/1/gate/openstack-tox-lower-constraints/b6867a6/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/96/668196/1/gate/openstack-tox-pep8/2eb529b/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/96/668196/1/gate/openstack-tox-py27/f09b410/ : SUCCESS in 4m 58s\n- openstack-tox-py36 http://logs.openstack.org/96/668196/1/gate/openstack-tox-py36/2f75f6f/ : SUCCESS in 3m 34s\n- openstack-tox-py37 http://logs.openstack.org/96/668196/1/gate/openstack-tox-py37/f4843c2/ : SUCCESS in 3m 34s\n- tempest-full http://logs.openstack.org/96/668196/1/gate/tempest-full/9752ccd/ : SUCCESS in 2h 02m 32s\n- tempest-full-py3 http://logs.openstack.org/96/668196/1/gate/tempest-full-py3/3454834/ : SUCCESS in 1h 42m 02s\n- build-openstack-releasenotes http://logs.openstack.org/96/668196/1/gate/build-openstack-releasenotes/fd72ddd/html/ : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7e54c8835d8b6ea544f4e1096db37524b915c313","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 22:09:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a58374fde64fdc14e327940dde6bea4a8a39345","revisions":{"7a58374fde64fdc14e327940dde6bea4a8a39345":{"kind":"REWORK","_number":1,"created":"2019-06-28 15:39:46.000000000","uploader":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"ref":"refs/changes/96/668196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/96/668196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/668196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/668196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/668196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/96/668196/1"}}},"commit":{"parents":[{"commit":"89b542f4fa9db27e663c0ac60a3ceb379ecf0663","subject":"Merge \"Update flake8 style enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/89b542f4fa9db27e663c0ac60a3ceb379ecf0663"}]}],"author":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-06-28 15:39:42.000000000","tz":-360},"committer":{"name":"Boden R","email":"bodenvmw@gmail.com","date":"2019-06-28 15:39:42.000000000","tz":-360},"subject":"make sql fixtures public","message":"make sql fixtures public\n\nThis patch makes the _StaticSqlFixture and _SqlFixture fixture classes\npublic by removing the leading underscore in the name. Based on\nrecent testing [1][2] the current versions of these fixtures seems to\nwork for consumers.\n\n[1] https://review.opendev.org/#/c/657887/\n[2] https://review.opendev.org/#/c/668000/\n\nChange-Id: I5e9497d650176201407f0823f0fce46e009fb89c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7a58374fde64fdc14e327940dde6bea4a8a39345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7a58374fde64fdc14e327940dde6bea4a8a39345"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
