)]}'
{"id":"openstack%2Fneutron-lib~670908","triplet_id":"openstack%2Fneutron-lib~master~I6b6c34c13f006ffca4cb4e2f18a2f825764374ec","project":"openstack/neutron-lib","branch":"master","topic":"bug/1582500","hashtags":[],"change_id":"I6b6c34c13f006ffca4cb4e2f18a2f825764374ec","subject":"Only have one number to name mapping for IPv6 ICMP","status":"MERGED","created":"2019-07-15 19:41:33.000000000","updated":"2019-07-16 10:16:37.000000000","submitted":"2019-07-16 10:16:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670908-1563272197749-e47bf6dd","meta_rev_id":"64553d7e458376cdf6625be282e7c98f2fe5d400","_number":670908,"virtual_id_number":670908,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-07-16 10:16:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2019-07-16 08:36:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-07-15 19:44:33.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2019-07-16 08:36:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 19:44:33.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-07-16 08:36:10.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-07-16 10:16:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48f98e3fd26a651a29456f20dc37600eafec2a30","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-15 19:41:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"72d83203fb369c4f6282a06533825d8cbc7ead67","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-07-15 19:44:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"424996094e22ab4544a2a8eff0e2ad669045ef8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 21:30:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/670908/1/check/openstack-tox-docs/798523d/html/ : SUCCESS in 3m 29s\n- openstack-tox-cover http://logs.openstack.org/08/670908/1/check/openstack-tox-cover/09374c5/cover/ : SUCCESS in 4m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/670908/1/check/openstack-tox-lower-constraints/84d0293/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 http://logs.openstack.org/08/670908/1/check/openstack-tox-pep8/a036798/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/08/670908/1/check/openstack-tox-py27/98ca9a2/ : SUCCESS in 3m 34s\n- openstack-tox-py36 http://logs.openstack.org/08/670908/1/check/openstack-tox-py36/3c98870/ : SUCCESS in 4m 38s\n- openstack-tox-py37 http://logs.openstack.org/08/670908/1/check/openstack-tox-py37/81eac99/ : SUCCESS in 4m 20s\n- tempest-full http://logs.openstack.org/08/670908/1/check/tempest-full/28428fd/ : SUCCESS in 1h 39m 07s\n- tempest-full-py3 http://logs.openstack.org/08/670908/1/check/tempest-full-py3/5c59cc3/ : SUCCESS in 1h 24m 31s","accounts_in_message":[],"_revision_number":1},{"id":"30daf67567de6f97cbc21647bbaca7f54b1b5955","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-16 08:36:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"746ccc3bfd08f5cf9722d93f4ae4115dc962db53","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 08:36:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10e084ae4616afc73eea298983909b1473e8fda9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 10:16:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/08/670908/1/gate/openstack-tox-docs/0c4947f/html/ : SUCCESS in 3m 02s\n- openstack-tox-lower-constraints http://logs.openstack.org/08/670908/1/gate/openstack-tox-lower-constraints/71b0582/ : SUCCESS in 4m 34s\n- openstack-tox-pep8 http://logs.openstack.org/08/670908/1/gate/openstack-tox-pep8/fecc42a/ : SUCCESS in 3m 48s\n- openstack-tox-py27 http://logs.openstack.org/08/670908/1/gate/openstack-tox-py27/fb7320d/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/08/670908/1/gate/openstack-tox-py36/d9a74e5/ : SUCCESS in 3m 55s\n- openstack-tox-py37 http://logs.openstack.org/08/670908/1/gate/openstack-tox-py37/f2bcecd/ : SUCCESS in 4m 14s\n- tempest-full http://logs.openstack.org/08/670908/1/gate/tempest-full/d7ea55d/ : SUCCESS in 1h 33m 39s\n- tempest-full-py3 http://logs.openstack.org/08/670908/1/gate/tempest-full-py3/32d648a/ : SUCCESS in 1h 15m 04s","accounts_in_message":[],"_revision_number":1},{"id":"bf7bf5416d936cae046e4351497d65ade3b23880","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 10:16:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1376d2e807dc020d027dc5fa07e317ab066eda75","revisions":{"1376d2e807dc020d027dc5fa07e317ab066eda75":{"kind":"REWORK","_number":1,"created":"2019-07-15 19:41:33.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/08/670908/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/08/670908/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/670908/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/670908/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/08/670908/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/08/670908/1"}}},"commit":{"parents":[{"commit":"7778e3118bc78fd4c6b88386567ab9d7523d1d54","subject":"Merge \"Add shim extension admin_state_down_before_update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7778e3118bc78fd4c6b88386567ab9d7523d1d54"}]}],"author":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-07-15 19:35:47.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"bhaley@redhat.com","date":"2019-07-15 19:35:47.000000000","tz":-240},"subject":"Only have one number to name mapping for IPv6 ICMP","message":"Only have one number to name mapping for IPv6 ICMP\n\nNeutron normalized the IPv6 ICMP protocol string to \u0027ipv6-icmp\u0027,\nbut the number to name mapping dict incorrectly shows \u0027icmpv6\u0027:\n\n\u003e\u003e\u003e constants.IP_PROTOCOL_NUM_TO_NAME_MAP[\u002758\u0027]\n\u0027icmpv6\u0027\n\nThis is because in the name to number mapping dict there are\ntwo keys, \u0027ipv6-icmp\u0027 and \u0027icmpv6\u0027, that have a value of 58\nin order to keep backwards-compatibility. Since the number\nto name mapping is built automatically from that, it overwrites\nthe \u002758\u0027 key value to be \u0027icmpv6\u0027. Filter it out so there is\nonly a single mapping to \u0027ipv6-icmp\u0027.\n\nChange-Id: I6b6c34c13f006ffca4cb4e2f18a2f825764374ec\nRelated-bug: #1582500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1376d2e807dc020d027dc5fa07e317ab066eda75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1376d2e807dc020d027dc5fa07e317ab066eda75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
