)]}'
{"id":"openstack%2Fneutron-lib~671528","triplet_id":"openstack%2Fneutron-lib~master~Idd9b442a30de294ce3fc97069cadba4457d746ab","project":"openstack/neutron-lib","branch":"master","topic":"l3-conntrack-helper","hashtags":[],"change_id":"Idd9b442a30de294ce3fc97069cadba4457d746ab","subject":"Fix - L3 Conntrack Helper validator","status":"MERGED","created":"2019-07-18 17:15:57.000000000","updated":"2019-07-19 17:07:02.000000000","submitted":"2019-07-19 17:07:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671528-1563556022240-b2961d89","meta_rev_id":"e6e393dcbadd4bb6df9b1582daa1893af2c368f9","_number":671528,"virtual_id_number":671528,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-07-19 17:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2019-07-19 08:01:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2019-07-19 07:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-07-19 15:03:01.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-07-19 15:03:01.000000000","_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 17:42:49.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-07-19 07:26:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-07-19 08:01:56.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-07-19 15:03:01.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-07-19 17:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e5e2dd6476ab8f8780aadb9416d37c860c4db25","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-07-18 17:15:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4110b100952f6fca51304127197e282598e017d","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-07-18 17:19:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b442edd9da95bfb1b809192de83661ddeffa145d","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-07-18 17:25:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"20fed4daba5c2ba2bb63c563c80b1bf2361b6ea0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-07-18 17:36:18.000000000","message":"Patch Set 3: Code-Review+1\n\nGood catch","accounts_in_message":[],"_revision_number":3},{"id":"cf7ed69da417de8e686529e1ed334798a5f23803","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-18 17:42:49.000000000","message":"Patch Set 3:\n\n(4 comments)\n\ntypo in release note","accounts_in_message":[],"_revision_number":3},{"id":"2c5c91be77be9300b2af766032309e59fa4c8d87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 19:25:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/671528/3/check/openstack-tox-docs/13cbd8d/html/ : SUCCESS in 3m 25s\n- openstack-tox-cover http://logs.openstack.org/28/671528/3/check/openstack-tox-cover/ad52dbf/cover/ : SUCCESS in 3m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/28/671528/3/check/openstack-tox-lower-constraints/edbc67b/ : SUCCESS in 4m 37s\n- openstack-tox-pep8 http://logs.openstack.org/28/671528/3/check/openstack-tox-pep8/72bcb8b/ : FAILURE in 3m 55s\n- openstack-tox-py27 http://logs.openstack.org/28/671528/3/check/openstack-tox-py27/573279e/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/28/671528/3/check/openstack-tox-py36/e35450c/ : SUCCESS in 5m 16s\n- openstack-tox-py37 http://logs.openstack.org/28/671528/3/check/openstack-tox-py37/3b25e55/ : SUCCESS in 5m 39s\n- tempest-full http://logs.openstack.org/28/671528/3/check/tempest-full/627838c/ : SUCCESS in 1h 51m 49s\n- tempest-full-py3 http://logs.openstack.org/28/671528/3/check/tempest-full-py3/39220cc/ : SUCCESS in 1h 23m 59s\n- build-openstack-releasenotes http://logs.openstack.org/28/671528/3/check/build-openstack-releasenotes/f9ca9d6/html/ : SUCCESS in 3m 03s","accounts_in_message":[],"_revision_number":3},{"id":"20024e1875b629be7e4a7562898b414ea6ea9387","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-07-18 21:46:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a01e21cb8d393a117bb6994dcc855ad1215bc253","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-07-18 21:48:36.000000000","message":"Patch Set 4:\n\n(4 comments)\n\nThanks Brian.","accounts_in_message":[],"_revision_number":4},{"id":"b33f49001fd88b76291982bb59499b6d374849d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 23:29:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/28/671528/4/check/openstack-tox-docs/ca8e071/html/ : SUCCESS in 2m 34s\n- openstack-tox-cover http://logs.openstack.org/28/671528/4/check/openstack-tox-cover/423812b/cover/ : SUCCESS in 3m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/28/671528/4/check/openstack-tox-lower-constraints/813f025/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/28/671528/4/check/openstack-tox-pep8/add1164/ : FAILURE in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/28/671528/4/check/openstack-tox-py27/e824dd3/ : SUCCESS in 3m 33s\n- openstack-tox-py36 http://logs.openstack.org/28/671528/4/check/openstack-tox-py36/914eab9/ : SUCCESS in 5m 14s\n- openstack-tox-py37 http://logs.openstack.org/28/671528/4/check/openstack-tox-py37/ed659be/ : SUCCESS in 4m 36s\n- tempest-full http://logs.openstack.org/28/671528/4/check/tempest-full/0377837/ : SUCCESS in 1h 38m 44s\n- tempest-full-py3 http://logs.openstack.org/28/671528/4/check/tempest-full-py3/708c921/ : SUCCESS in 1h 31m 04s\n- build-openstack-releasenotes http://logs.openstack.org/28/671528/4/check/build-openstack-releasenotes/5edd884/html/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":4},{"id":"a98303653c1f7ad5502efb7a27d961130a6e51e4","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2019-07-19 00:17:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ca53bcb2d6e3b7233fa9baf32c28a5128adb8438","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 02:00:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/28/671528/5/check/openstack-tox-docs/f677e7d/html/ : SUCCESS in 2m 43s\n- openstack-tox-cover http://logs.openstack.org/28/671528/5/check/openstack-tox-cover/18032fd/cover/ : SUCCESS in 3m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/28/671528/5/check/openstack-tox-lower-constraints/082b90e/ : SUCCESS in 4m 26s\n- openstack-tox-pep8 http://logs.openstack.org/28/671528/5/check/openstack-tox-pep8/b4f2883/ : SUCCESS in 3m 29s\n- openstack-tox-py27 http://logs.openstack.org/28/671528/5/check/openstack-tox-py27/6c3d767/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/28/671528/5/check/openstack-tox-py36/d8f74a6/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/28/671528/5/check/openstack-tox-py37/86b0b5e/ : SUCCESS in 3m 41s\n- tempest-full http://logs.openstack.org/28/671528/5/check/tempest-full/312e10a/ : SUCCESS in 1h 40m 57s\n- tempest-full-py3 http://logs.openstack.org/28/671528/5/check/tempest-full-py3/52a7266/ : SUCCESS in 1h 24m 33s\n- build-openstack-releasenotes http://logs.openstack.org/28/671528/5/check/build-openstack-releasenotes/77f1a38/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":5},{"id":"c223b4868fcb8a9208aab47abab828b90ba3614b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-19 07:26:47.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bf8249c2c1009f2bd76db72ed1a3e4827c0f1de1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-07-19 08:01:56.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"fbcf59d94f64846e75a1a9329ef6bd2a5db297c9","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-07-19 15:03:01.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7e74ce20884bdba47721ed85ab59ade4f2d40ee4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 15:07:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c61a0e588a2c3fcee587e0f3ffec93d40146b0dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:07:01.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/28/671528/5/gate/openstack-tox-docs/9f932f8/html/ : SUCCESS in 2m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/28/671528/5/gate/openstack-tox-lower-constraints/3cab384/ : SUCCESS in 4m 29s\n- openstack-tox-pep8 http://logs.openstack.org/28/671528/5/gate/openstack-tox-pep8/d40553a/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/28/671528/5/gate/openstack-tox-py27/8669f05/ : SUCCESS in 3m 42s\n- openstack-tox-py36 http://logs.openstack.org/28/671528/5/gate/openstack-tox-py36/6f7b019/ : SUCCESS in 3m 25s\n- openstack-tox-py37 http://logs.openstack.org/28/671528/5/gate/openstack-tox-py37/909f1a3/ : SUCCESS in 3m 51s\n- tempest-full http://logs.openstack.org/28/671528/5/gate/tempest-full/e7ccc77/ : SUCCESS in 1h 57m 33s\n- tempest-full-py3 http://logs.openstack.org/28/671528/5/gate/tempest-full-py3/07dc539/ : SUCCESS in 1h 20m 04s\n- build-openstack-releasenotes http://logs.openstack.org/28/671528/5/gate/build-openstack-releasenotes/f95cdba/html/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":5},{"id":"b6cc0bc6cdc46478eba395e3bfa19ced525d8671","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 17:07:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6d9f9e2841edcf3a92bbcba7f7a45d9d464a14ee","revisions":{"b53b84d07f77a8344b67665e57b66fc46ce5001d":{"kind":"REWORK","_number":1,"created":"2019-07-18 17:15:57.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/28/671528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/671528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/1"}}},"commit":{"parents":[{"commit":"1376d2e807dc020d027dc5fa07e317ab066eda75","subject":"Only have one number to name mapping for IPv6 ICMP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1376d2e807dc020d027dc5fa07e317ab066eda75"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-18 17:08:51.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-18 17:13:27.000000000","tz":120},"subject":"Fix - L3 Conntrack Helper validator","message":"Fix - L3 Conntrack Helper validator\n\n\u0027type:string\u0027 is validated by lenght, the defenition was\npassing in validators.validate_string instead of a number.\nIn python3 this causes a TypeError:\n \u0027\u003e\u0027 not supported between instances of \u0027int\u0027 and \u0027function\u0027\n\nSet the length to 64, matching the db model:\nhttps://opendev.org/openstack/neutron/src/branch/master/neutron/db/models/conntrack_helper.py#L32\n\nRelated-Bug: #1823633\nChange-Id: Idd9b442a30de294ce3fc97069cadba4457d746ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b53b84d07f77a8344b67665e57b66fc46ce5001d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b53b84d07f77a8344b67665e57b66fc46ce5001d"}]},"branch":"refs/heads/master"},"c3a3d700721aa678367bc77ae749a84480ec47b5":{"kind":"REWORK","_number":2,"created":"2019-07-18 17:19:59.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/28/671528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/671528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/2"}}},"commit":{"parents":[{"commit":"1376d2e807dc020d027dc5fa07e317ab066eda75","subject":"Only have one number to name mapping for IPv6 ICMP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1376d2e807dc020d027dc5fa07e317ab066eda75"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-18 17:08:51.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-18 17:18:44.000000000","tz":120},"subject":"Fix - L3 Conntrack Helper validator","message":"Fix - L3 Conntrack Helper validator\n\n\u0027type:string\u0027 is validated by lenght, the defenition was\npassing in validators.validate_string instead of a number.\nIn python3 this causes a TypeError:\n \u0027\u003e\u0027 not supported between instances of \u0027int\u0027 and \u0027function\u0027\n\nSet the length to 64, matching the db model:\nhttps://opendev.org/openstack/neutron/src/branch/master/neutron/db/models/conntrack_helper.py#L32\n\nRelated-Bug: #1823633\nChange-Id: Idd9b442a30de294ce3fc97069cadba4457d746ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c3a3d700721aa678367bc77ae749a84480ec47b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c3a3d700721aa678367bc77ae749a84480ec47b5"}]},"branch":"refs/heads/master"},"db9df87960a8c1abad54b7d192069ea1f0cc0c46":{"kind":"REWORK","_number":3,"created":"2019-07-18 17:25:30.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/28/671528/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/671528/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/3"}}},"commit":{"parents":[{"commit":"1376d2e807dc020d027dc5fa07e317ab066eda75","subject":"Only have one number to name mapping for IPv6 ICMP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1376d2e807dc020d027dc5fa07e317ab066eda75"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-18 17:08:51.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-18 17:25:19.000000000","tz":120},"subject":"Fix - L3 Conntrack Helper validator","message":"Fix - L3 Conntrack Helper validator\n\n\u0027type:string\u0027 is validated by lenght, the defenition was\npassing in validators.validate_string instead of a number.\nIn python3 this causes a TypeError:\n \u0027\u003e\u0027 not supported between instances of \u0027int\u0027 and \u0027function\u0027\n\nSet the length to 64, matching the db model:\nhttps://opendev.org/openstack/neutron/src/branch/master/neutron/db/models/conntrack_helper.py#L32\n\nRelated-Bug: #1823633\nChange-Id: Idd9b442a30de294ce3fc97069cadba4457d746ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/db9df87960a8c1abad54b7d192069ea1f0cc0c46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/db9df87960a8c1abad54b7d192069ea1f0cc0c46"}]},"branch":"refs/heads/master"},"60cee9c12e6917fbe4b5d20bfaf51256c91e42bf":{"kind":"REWORK","_number":4,"created":"2019-07-18 21:46:58.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/28/671528/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/671528/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/4"}}},"commit":{"parents":[{"commit":"1376d2e807dc020d027dc5fa07e317ab066eda75","subject":"Only have one number to name mapping for IPv6 ICMP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1376d2e807dc020d027dc5fa07e317ab066eda75"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-18 17:08:51.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-18 21:44:23.000000000","tz":120},"subject":"Fix - L3 Conntrack Helper validator","message":"Fix - L3 Conntrack Helper validator\n\n\u0027type:string\u0027 is validated by length, the definition was\npassing in validators.validate_string instead of a number.\nIn python3 this causes a TypeError:\n \u0027\u003e\u0027 not supported between instances of \u0027int\u0027 and \u0027function\u0027\n\nSet the length to 64, matching the db model:\nhttps://opendev.org/openstack/neutron/src/branch/master/neutron/db/models/conntrack_helper.py#L32\n\nRelated-Bug: #1823633\nChange-Id: Idd9b442a30de294ce3fc97069cadba4457d746ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/60cee9c12e6917fbe4b5d20bfaf51256c91e42bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/60cee9c12e6917fbe4b5d20bfaf51256c91e42bf"}]},"branch":"refs/heads/master"},"6d9f9e2841edcf3a92bbcba7f7a45d9d464a14ee":{"kind":"REWORK","_number":5,"created":"2019-07-19 00:17:23.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/28/671528/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/28/671528/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/28/671528/5"}}},"commit":{"parents":[{"commit":"1376d2e807dc020d027dc5fa07e317ab066eda75","subject":"Only have one number to name mapping for IPv6 ICMP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/1376d2e807dc020d027dc5fa07e317ab066eda75"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-18 17:08:51.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2019-07-19 00:15:58.000000000","tz":120},"subject":"Fix - L3 Conntrack Helper validator","message":"Fix - L3 Conntrack Helper validator\n\n\u0027type:string\u0027 is validated by length, the definition was\npassing in validators.validate_string instead of a number.\nIn python3 this causes a TypeError:\n \u0027\u003e\u0027 not supported between instances of \u0027int\u0027 and \u0027function\u0027\n\nSet the length to 64, matching the db model:\nhttps://opendev.org/openstack/neutron/src/branch/master/neutron/db/models/conntrack_helper.py#L32\n\nRelated-Bug: #1823633\nChange-Id: Idd9b442a30de294ce3fc97069cadba4457d746ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6d9f9e2841edcf3a92bbcba7f7a45d9d464a14ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6d9f9e2841edcf3a92bbcba7f7a45d9d464a14ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
