)]}'
{"id":"openstack%2Fneutron-lib~673589","triplet_id":"openstack%2Fneutron-lib~master~Ia756506e25d7b74030f55d1032b5cb71f5e1925b","project":"openstack/neutron-lib","branch":"master","topic":"retire-api-site","hashtags":[],"change_id":"Ia756506e25d7b74030f55d1032b5cb71f5e1925b","subject":"Update api-ref location","status":"MERGED","created":"2019-07-30 17:41:22.000000000","updated":"2019-07-31 10:03:54.000000000","submitted":"2019-07-31 10:03:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673589-1564567434370-4b3fabf7","meta_rev_id":"eec2d6121ee5898d12110d535022fcf596a1ecd7","_number":673589,"virtual_id_number":673589,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2019-07-31 10:03:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":2,"date":"2019-07-31 08:22:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2019-07-31 02:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-31 08:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":1,"date":"2019-07-31 08:22:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 02:14:13.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-07-31 08:18:29.000000000","updated_by":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"reviewer":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"state":"REVIEWER"},{"updated":"2019-07-31 08:22:04.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-07-31 10:03:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31bcedc66fb4f225d4fbabd180ab4782eab405ff","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-07-30 17:41:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c257833b514813c77c8ab7ff076d7386fda47620","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 19:22:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/89/673589/1/check/build-openstack-api-ref/26283ae/html/ : SUCCESS in 4m 13s\n- openstack-tox-docs http://logs.openstack.org/89/673589/1/check/openstack-tox-docs/66e0915/html/ : SUCCESS in 3m 47s\n- openstack-tox-cover http://logs.openstack.org/89/673589/1/check/openstack-tox-cover/ff6b5bb/cover/ : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/673589/1/check/openstack-tox-lower-constraints/bd90559/ : SUCCESS in 4m 50s\n- openstack-tox-pep8 http://logs.openstack.org/89/673589/1/check/openstack-tox-pep8/6d32964/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/89/673589/1/check/openstack-tox-py27/ad60796/ : SUCCESS in 3m 15s\n- openstack-tox-py36 http://logs.openstack.org/89/673589/1/check/openstack-tox-py36/97425e4/ : SUCCESS in 3m 58s\n- openstack-tox-py37 http://logs.openstack.org/89/673589/1/check/openstack-tox-py37/a25afd3/ : SUCCESS in 4m 36s\n- tempest-full http://logs.openstack.org/89/673589/1/check/tempest-full/a9d097b/ : SUCCESS in 1h 23m 05s\n- tempest-full-py3 http://logs.openstack.org/89/673589/1/check/tempest-full-py3/ca2a60f/ : SUCCESS in 1h 30m 49s","accounts_in_message":[],"_revision_number":1},{"id":"14281b70de86d5038116b556256c56f4bd255f39","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-07-31 02:14:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d9521bc869ca7b83750974fe838a89e2fe268fdf","author":{"_account_id":27822,"name":"pengyuesheng","email":"pengyuesheng@gohighsec.com","username":"pengyuesheng"},"date":"2019-07-31 08:18:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a43f1ce13018daa1ab5c6cc55ab8db462edee74f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-07-31 08:22:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2d2079416786c849e3bc40ce35cf7720154bf46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 08:22:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7dbc5ac510aaf52146cc15c3d0e5c0c776f6d4d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 10:03:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref http://logs.openstack.org/89/673589/1/gate/build-openstack-api-ref/191a9da/html/ : SUCCESS in 4m 09s\n- openstack-tox-docs http://logs.openstack.org/89/673589/1/gate/openstack-tox-docs/43460a0/html/ : SUCCESS in 3m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/673589/1/gate/openstack-tox-lower-constraints/4450fa9/ : SUCCESS in 4m 14s\n- openstack-tox-pep8 http://logs.openstack.org/89/673589/1/gate/openstack-tox-pep8/116d76a/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/89/673589/1/gate/openstack-tox-py27/a18084e/ : SUCCESS in 6m 18s\n- openstack-tox-py36 http://logs.openstack.org/89/673589/1/gate/openstack-tox-py36/a94c623/ : SUCCESS in 5m 06s\n- openstack-tox-py37 http://logs.openstack.org/89/673589/1/gate/openstack-tox-py37/8ab531d/ : SUCCESS in 4m 39s\n- tempest-full http://logs.openstack.org/89/673589/1/gate/tempest-full/44b63db/ : SUCCESS in 1h 38m 13s\n- tempest-full-py3 http://logs.openstack.org/89/673589/1/gate/tempest-full-py3/92bbc3c/ : SUCCESS in 1h 26m 29s","accounts_in_message":[],"_revision_number":1},{"id":"42c696984dbfa2f1b90e154eede1f31da0283a91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 10:03:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d27313f42588c397fd8d8f6d84e199404385eb54","revisions":{"d27313f42588c397fd8d8f6d84e199404385eb54":{"kind":"REWORK","_number":1,"created":"2019-07-30 17:41:22.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/89/673589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/89/673589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/673589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/673589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/89/673589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/89/673589/1"}}},"commit":{"parents":[{"commit":"aaae83acaac5f58918ff53613b4de20bbfe30d81","subject":"Mention that list of e.g. IDs is supported in GET requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/aaae83acaac5f58918ff53613b4de20bbfe30d81"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-30 17:41:11.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-07-30 17:41:11.000000000","tz":120},"subject":"Update api-ref location","message":"Update api-ref location\n\nThe api documentation is now published on docs.openstack.org instead\nof developer.openstack.org. Update all links that are changed to the\nnew location.\n\nNote that redirects will be set up as well but let\u0027s point now to the\nnew location.\n\nFor details, see:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-July/007828.html\n\nChange-Id: Ia756506e25d7b74030f55d1032b5cb71f5e1925b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d27313f42588c397fd8d8f6d84e199404385eb54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d27313f42588c397fd8d8f6d84e199404385eb54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
