)]}'
{"id":"openstack%2Fneutron-lib~679259","triplet_id":"openstack%2Fneutron-lib~master~I95927d61a3590ea3495bc2716c1fa5fd4feb2948","project":"openstack/neutron-lib","branch":"master","topic":"extraroute","hashtags":[],"change_id":"I95927d61a3590ea3495bc2716c1fa5fd4feb2948","subject":"Introduce new member actions additively","status":"MERGED","created":"2019-08-29 14:00:04.000000000","updated":"2019-08-30 10:02:08.000000000","submitted":"2019-08-30 10:00:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679259-1567159232402-6ccbb2a0","meta_rev_id":"276f45d9c7c0e3947b18b91c0dcf5593bb4315c9","_number":679259,"virtual_id_number":679259,"owner":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2019-08-30 10:00:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":2,"date":"2019-08-30 00:28:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":2,"date":"2019-08-30 08:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-08-30 05:03:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":1,"date":"2019-08-30 08:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-29 14:19:04.000000000","updated_by":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"reviewer":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"state":"REVIEWER"},{"updated":"2019-08-29 14:36:12.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-08-30 00:28:08.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-08-30 05:03:53.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-08-30 08:18:43.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-08-30 10:00:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cdebad5413cbd2b5d11262d1e121ac6936e58198","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-08-29 14:00:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0416058accc8952e5b4dfb6449add834d3073d5","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-08-29 14:17:40.000000000","message":"Patch Set 1:\n\nThis change is meant to work around the functional test failure encountered in the neutron implementation:\n\nhttps://review.opendev.org/670851 (see patch set 6)\n\nIf you find this approach okay, we may want to get this merged before the final neutron-lib for train is released.","accounts_in_message":[],"_revision_number":1},{"id":"516f59e50c85f4eed4b72442264f8fae11661c5f","author":{"_account_id":5367,"name":"boden","email":"bodenvmw@gmail.com","username":"boden"},"date":"2019-08-29 14:19:04.000000000","message":"Patch Set 1:\n\nSeems reasonable; are we sure this will address the issue without any regressions?\n\nAlso; does it make sense to add a release note for tracking purposes?","accounts_in_message":[],"_revision_number":1},{"id":"f73bc197a9599371bc1d5ef452c27d82111664d9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-29 14:34:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4fef4ad7b6b8d62a2be7458f2d117c1439bf28cf","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-08-29 14:35:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f7f41f13726fb7a5b4c0d8aaf3a5e75c28b2a7b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-08-29 14:36:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d59f9567c1ff9f323243959386c5039d3fe691d0","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-08-29 15:09:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d9359e2838c03f2cdc909892d109d61182cd68b","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2019-08-29 15:10:58.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks everyone for the quick reviews! Added a release note too.","accounts_in_message":[],"_revision_number":1},{"id":"dffced321d8e3d2d256bb831ba11ffab754fd90e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 17:15:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/607013f84489400ea33c02b574d4dce8 : SUCCESS in 6m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56cee36bf9f944cf9b21172ba97c408a : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fbf23a74adf4debb25a28bc426d8dc0 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/782d99537aad45a18118128bfde2990d : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6b32f98f5734b44914eb4e38a839243 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abddc9254ee84432ad33d07339bf55a8 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba833abb0e4249848567cc24b3651529 : SUCCESS in 5m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4e590b2af5994deebca5bcbd9b2e33b2 : SUCCESS in 2h 00m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/649da637c25b426e98dbb344ca34ce6c : SUCCESS in 1h 27m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83fad5652b0b4a65b9d7c1c7501d3198 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"07fdc9d7dd9c0d41428fd2610f65061b3457b9e2","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-08-30 00:27:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8b87d681cc41cc50ee099ac66bfa2c34660714b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-08-30 00:27:43.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1554f7ccf515830e72113db00d5966dcab25dee9","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-08-30 00:28:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"7d4f4fbf06b1bcceaab3eacaaa5c051476c8d2a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 02:21:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e20119e19b04b828503398629ad49f8 : SUCCESS in 5m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef518167fd654fb2b38c2e047bc1c667 : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d288775b5fc548ee8344452d0814df2b : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19c411deb10545079b9c5a6d01a974b4 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3963abb4ca8444f2a3995b720ae533df : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11ccecd4c71b4b1bb7baa67f93282b68 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e388c51daef41458bed42e0aaaa9aee : SUCCESS in 5m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e00f5042e75f45eb8fc8300827bfbcf6 : SUCCESS in 1h 44m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c379736c680c4b969c130dc0cfcc665f : SUCCESS in 1h 27m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3c3582f4b9b4702b4a235b703e01f45 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":3},{"id":"895e1a5f6863fbbbea51a0e2cd661216d0718d40","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-08-30 05:03:53.000000000","message":"Patch Set 3: Code-Review+1\n\nGood to learn again something :-)","accounts_in_message":[],"_revision_number":3},{"id":"bdaec4027df08c914a57eba64fb05b29cbca751d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-08-30 08:18:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nI tested locally patch https://review.opendev.org/#/c/670851/ with this one and broken test is fixed.","accounts_in_message":[],"_revision_number":3},{"id":"2810bd18182c4e944950fa660b23d6879c15d6ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 08:18:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6ff35fc268456221f7d4a0e797548238488c3d9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 10:00:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32adcb14d26c4209afe5e0cae231a469 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cdebce7a2c15422da23f57810ddb4f2d : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c365e396b5421da4fa7eff59312d19 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf8255af27dd43b782231aff73664b6f : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95394fa5fdf3482bb094b0e660576315 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7dfb4ab430304d169a01ad1317424724 : SUCCESS in 4m 36s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f1000cc0a6434ba38686620c01b6a789 : SUCCESS in 1h 39m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8975d9bd5f84c85aec57d8f414eeadf : SUCCESS in 1h 28m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a97a3c5820164f78a916fc0b01c9090d : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"d4d6b6a3c937378ae8440a45df3b6768c2a4f7b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 10:00:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ffc8b3732b38a0905fd0c7e65a13f7efcbc3e151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 10:02:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b64d1119ba64e56b235ba38d92848fa : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4ef4b711b75e5329e09ed5eedca36dea35445494","revisions":{"4294d26e1b5309c8d2c6e7fe14fbfca36cf41fdd":{"kind":"REWORK","_number":1,"created":"2019-08-29 14:00:04.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/59/679259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/59/679259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/1"}}},"commit":{"parents":[{"commit":"104627a9336c2cff6808dc0024c43c6b0a7a723c","subject":"New api-def: extraroute-atomic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/104627a9336c2cff6808dc0024c43c6b0a7a723c"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2019-08-29 12:15:37.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2019-08-29 12:48:38.000000000","tz":120},"subject":"Work around the request router\u0027s limitation","message":"Work around the request router\u0027s limitation\n\n... of not being able to fully process member actions for the same\nresource coming from multiple api-defs.\n\nIn the \u0027extraroute-atomic\u0027 api-def as it was first merged [1]\nsome router member actions (PUT add/remove_extraroutes) were defined\nin one api-def (extraroute_atomic.py), while others\n(PUT add/remove_router_interface) were defined in another (l3.py).\n\nThis (in the subsequent neutron implementation change [2]) broke\nfunctional test:\n\nneutron.tests.functional.pecan_wsgi.test_controllers.\\\nTestRouterController.test_unsupported_method_member_action\n\nThis test sent POST/GET /v2.0/routers/router-id/add_router_interface,\nand expected response 405 Method Not Allowed. Due to the split\ndefinition it actually received 404 Not Found. In the meanwhile all\nfour PUT member actions were working properly.\n\nAt the moment I don\u0027t see a better fix than to move all member actions\nto l3.py and turn extraroute_atomic.py into a shim extension.\n\n[1] https://review.opendev.org/670849\n[2] https://review.opendev.org/#/c/670851/6\n\nChange-Id: I95927d61a3590ea3495bc2716c1fa5fd4feb2948\nPartial-Bug: #1826396 (rfe)\nRelated-Change: https://review.opendev.org/655680 (spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4294d26e1b5309c8d2c6e7fe14fbfca36cf41fdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4294d26e1b5309c8d2c6e7fe14fbfca36cf41fdd"}]},"branch":"refs/heads/master"},"d68cda841b6894575b11f8cd35c9efc5ba0c93cb":{"kind":"REWORK","_number":2,"created":"2019-08-29 15:09:58.000000000","uploader":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"ref":"refs/changes/59/679259/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/59/679259/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/2"}}},"commit":{"parents":[{"commit":"104627a9336c2cff6808dc0024c43c6b0a7a723c","subject":"New api-def: extraroute-atomic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/104627a9336c2cff6808dc0024c43c6b0a7a723c"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2019-08-29 12:15:37.000000000","tz":120},"committer":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2019-08-29 14:53:33.000000000","tz":120},"subject":"Introduce new member actions additively","message":"Introduce new member actions additively\n\nIn the \u0027extraroute-atomic\u0027 api-def as it was first merged [1]\nsome router member actions (PUT add/remove_extraroutes) were defined\nin one api-def (extraroute_atomic.py), while others\n(PUT add/remove_router_interface) were defined in another (l3.py).\n\nThis (in the subsequent neutron implementation change [2]) broke\nfunctional test:\n\nneutron.tests.functional.pecan_wsgi.test_controllers.\\\nTestRouterController.test_unsupported_method_member_action\n\nThis test sent POST/GET /v2.0/routers/router-id/add_router_interface,\nand expected response 405 Method Not Allowed. Due to the split\ndefinition it actually received 404 Not Found. In the meanwhile all\nfour PUT member actions were working properly.\n\nThis change sets the \u0027extraroute-atomic\u0027 api-def\u0027s ACTION_MAP to contain\nthe member actions of both the \u0027router\u0027 and the \u0027extraroute-atomic\u0027\nextensions\u0027 member actions in order to get all four member actions\nproperly defined.\n\n[1] https://review.opendev.org/670849\n[2] https://review.opendev.org/#/c/670851/6\n\nChange-Id: I95927d61a3590ea3495bc2716c1fa5fd4feb2948\nPartial-Bug: #1826396 (rfe)\nRelated-Change: https://review.opendev.org/655680 (spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d68cda841b6894575b11f8cd35c9efc5ba0c93cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d68cda841b6894575b11f8cd35c9efc5ba0c93cb"}]},"branch":"refs/heads/master"},"4ef4b711b75e5329e09ed5eedca36dea35445494":{"kind":"REWORK","_number":3,"created":"2019-08-30 00:27:43.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/59/679259/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/59/679259/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/59/679259/3"}}},"commit":{"parents":[{"commit":"104627a9336c2cff6808dc0024c43c6b0a7a723c","subject":"New api-def: extraroute-atomic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/104627a9336c2cff6808dc0024c43c6b0a7a723c"}]}],"author":{"name":"Bence Romsics","email":"bence.romsics@gmail.com","date":"2019-08-29 12:15:37.000000000","tz":120},"committer":{"name":"Miguel Lavalle","email":"miguel.lavalle@verizonmedia.com","date":"2019-08-30 00:27:35.000000000","tz":0},"subject":"Introduce new member actions additively","message":"Introduce new member actions additively\n\nIn the \u0027extraroute-atomic\u0027 api-def as it was first merged [1]\nsome router member actions (PUT add/remove_extraroutes) were defined\nin one api-def (extraroute_atomic.py), while others\n(PUT add/remove_router_interface) were defined in another (l3.py).\n\nThis (in the subsequent neutron implementation change [2]) broke\nfunctional test:\n\nneutron.tests.functional.pecan_wsgi.test_controllers.\\\nTestRouterController.test_unsupported_method_member_action\n\nThis test sent POST/GET /v2.0/routers/router-id/add_router_interface,\nand expected response 405 Method Not Allowed. Due to the split\ndefinition it actually received 404 Not Found. In the meanwhile all\nfour PUT member actions were working properly.\n\nThis change sets the \u0027extraroute-atomic\u0027 api-def\u0027s ACTION_MAP to contain\nthe member actions of both the \u0027router\u0027 and the \u0027extraroute-atomic\u0027\nextensions\u0027 member actions in order to get all four member actions\nproperly defined.\n\n[1] https://review.opendev.org/670849\n[2] https://review.opendev.org/#/c/670851/6\n\nChange-Id: I95927d61a3590ea3495bc2716c1fa5fd4feb2948\nPartial-Bug: #1826396 (rfe)\nRelated-Change: https://review.opendev.org/655680 (spec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4ef4b711b75e5329e09ed5eedca36dea35445494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/4ef4b711b75e5329e09ed5eedca36dea35445494"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
