)]}'
{"id":"openstack%2Fneutron-lib~682279","triplet_id":"openstack%2Fneutron-lib~master~I3a8cb5be5c154577074d799e58a78ed10d33d8f7","project":"openstack/neutron-lib","branch":"master","hashtags":[],"change_id":"I3a8cb5be5c154577074d799e58a78ed10d33d8f7","subject":"update lbass api-ref: add health monitor type ``PING``","status":"ABANDONED","created":"2019-09-16 06:57:05.000000000","updated":"2019-09-16 09:29:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fef9eb8af11e9f5dfd6b72cb2efbfcd657e4aba3","_number":682279,"virtual_id_number":682279,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"date":"2019-09-16 09:28:57.000000000","_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":1,"date":"2019-09-16 07:08:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":-2,"date":"2019-09-16 09:12:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 07:08:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-16 09:12:15.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"5def00742e17af03f19d4b907e66772583286eea","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-09-16 06:57:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09e4ecc67a033f331028bde7538eefa0118d662b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 07:08:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b59262f0356f4f4faaafcd973e3dd5d9 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0277426dcff49dc989269325ec64f02 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac937666a4e4a2f9a008cef8e3c19ba : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5111d7a439cf3c21d026248104f8035f4e3a0356","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-09-16 09:12:15.000000000","message":"Patch Set 1: Code-Review-2\n\nThis is not needed as patch https://review.opendev.org/#/c/665828/ completly removes this docs due to removal of LBaaSv2","accounts_in_message":[],"_revision_number":1},{"id":"bff8e2745698cef5bcdca13e3590d90c4cfc84af","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-09-16 09:28:57.000000000","message":"Patch Set 1:\n\n\u003e This is not needed as patch https://review.opendev.org/#/c/665828/\n \u003e completly removes this docs due to removal of LBaaSv2\n\nthanks Slawek.I didn\u0027t notice that I would abandon it.","accounts_in_message":[],"_revision_number":1},{"id":"fef9eb8af11e9f5dfd6b72cb2efbfcd657e4aba3","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2019-09-16 09:29:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba767e22d866f3ff41131622d50bfea08c102a23","revisions":{"ba767e22d866f3ff41131622d50bfea08c102a23":{"kind":"REWORK","_number":1,"created":"2019-09-16 06:57:05.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/79/682279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/79/682279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/682279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/682279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/79/682279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/79/682279/1"}}},"commit":{"parents":[{"commit":"d800a41f669cb4aa3f949fc73fab50fda52b82a0","subject":"Merge \"PDF documentation build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/d800a41f669cb4aa3f949fc73fab50fda52b82a0"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-09-16 06:55:26.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2019-09-16 06:56:21.000000000","tz":480},"subject":"update lbass api-ref: add health monitor type ``PING``","message":"update lbass api-ref: add health monitor type ``PING``\n\nChange-Id: I3a8cb5be5c154577074d799e58a78ed10d33d8f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ba767e22d866f3ff41131622d50bfea08c102a23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ba767e22d866f3ff41131622d50bfea08c102a23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
