)]}'
{"id":"openstack%2Fneutron-lib~686097","triplet_id":"openstack%2Fneutron-lib~master~I6a1b983cf1bb0d6d04ec75b4725f39f29dc4c2ba","project":"openstack/neutron-lib","branch":"master","hashtags":[],"change_id":"I6a1b983cf1bb0d6d04ec75b4725f39f29dc4c2ba","subject":"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label","status":"MERGED","created":"2019-10-02 07:50:52.000000000","updated":"2019-10-27 11:51:19.000000000","submitted":"2019-10-27 11:49:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686097-1572176996439-cc9bba01","meta_rev_id":"cc7c725b120c17feb457ef3da085077a36e91edb","_number":686097,"virtual_id_number":686097,"owner":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30521,"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-27 11:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},{"value":2,"date":"2019-10-26 00:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-10-27 10:15:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30521,"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2019-10-27 10:15:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30521,"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":30521,"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30521,"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com"},{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-02 09:07:46.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-10-21 08:34:40.000000000","updated_by":{"_account_id":30521,"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com"},"reviewer":{"_account_id":30521,"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com"},"state":"REVIEWER"},{"updated":"2019-10-26 00:19:48.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-10-27 10:15:03.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-10-27 11:49:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10fd2e68f6466af1a9631929c556521f38a1a36b","author":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"date":"2019-10-02 07:50:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d41eb03aee27d71c828bd9b1b0e063c52aa84c6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-02 09:07:46.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nYou need some UTs for this.","accounts_in_message":[],"_revision_number":1},{"id":"0872470171ea885407191b7f11d881cbd9337df6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-02 09:54:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/424618c7068c4e29a07650108051d238 : SUCCESS in 5m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32a1222f324a468e989e5d091c569e35 : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b64de4a1deae482a99e8fb3656d223df : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4813621725b4da5b73128afff2f8a28 : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23011483c85f41e59f5fdb46ed297cd4 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7ada453d24e432fa379b4db292ce950 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f74d6ae86872400e9bb19ebe0a4d4e6f : SUCCESS in 4m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0af11ec429bb4fae847ead29c3aceb67 : SUCCESS in 1h 33m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f805c89b92e6449ba78f0669d0688501 : SUCCESS in 1h 56m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cc4bb8d993ca071a9760ec821b39a14bdd07f0b7","author":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"date":"2019-10-03 07:45:15.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"beaeac17006aaba94a06094696a1a7196eef974a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 09:56:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ff851bbe104b2d8a2d186d54ea4495 : SUCCESS in 4m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dbd0adcbb504e359ed50bdde8b6e50e : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af000d06d81843edb42c01e102286699 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/503c3547e817408bbf908181ba51fa6e : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e66f730c5e0843cba98e0ac86b647445 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3960ab50c80447d698a8b428f9143df8 : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cbaf6c783ce42b38f5df7d8309f11f3 : SUCCESS in 3m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0269ad1803644b8c9f44c46e0b72d516 : SUCCESS in 2h 10m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe6daee2dd5e4215af87a29c4bd2f850 : SUCCESS in 1h 17m 00s","accounts_in_message":[],"_revision_number":2},{"id":"61b31d3b1af1e8783e8150332aac2c9da883d77f","author":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"date":"2019-10-03 10:35:20.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dac4ff4ea1b600bfa8c975ad7608a5ef4fd9a843","author":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"date":"2019-10-21 06:43:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0cf42fe0cbe4138b3fb4bc7a220ef9e95e09c521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 08:07:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9bf92c4be9413c9228069b514a4e52 : SUCCESS in 5m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aeb3fa2231974f1cbd4fdecb6884a8a7 : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/acc69995aff347bb8f7c309ed087f494 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d76500d0494ffb800ba24e6d769b93 : FAILURE in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3fa6fa32d6d421080a3d48044220190 : FAILURE in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f75696ccd91a4765b51bfd9ef97a01aa : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ee400731f724b6da55c805adf870a8a : SUCCESS in 4m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/d484aa7675264e359a56685ec87642a7 : FAILURE in 21m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70c238bca2cb4771aa148980fa37f5cd : SUCCESS in 1h 21m 07s","accounts_in_message":[],"_revision_number":3},{"id":"e8da586a44f46822997767f1acbec5c67e10f888","author":{"_account_id":30521,"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com"},"date":"2019-10-21 08:34:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"bbda2e574a66e2a5349bbb41f2863074115c85a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 10:04:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33255227dde2410eb9e6a1bfc718de5e : SUCCESS in 4m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9cc8b5b9e6e43a7b92d2a76f7cc60e9 : SUCCESS in 3m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6189440e6d104bfb865861b710fec14a : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e0c9f8fca94991a4d4b68b94370073 : FAILURE in 2m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9c6dc899a7f4790812c5521d507ba90 : FAILURE in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94be48ca62e2488a904f5618e2e04e0a : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/25c0f2e739f04fc4b12af424f0b48980 : SUCCESS in 4m 05s\n- tempest-full https://zuul.opendev.org/t/openstack/build/67c5a0ad9be247e3a68c70d8790955b8 : FAILURE in 26m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64a71211296c4ef8942b8eff20195b87 : SUCCESS in 1h 24m 16s","accounts_in_message":[],"_revision_number":3},{"id":"3de615394a82fe7e43f2c5d0a7c27871e6c91956","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-22 12:35:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f4e13261c0a1040f24c146f0b54001287f301ca8","author":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"date":"2019-10-23 06:00:00.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"569d74c91e03fa400b0c0e742c77cc45460b392d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 08:14:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c4bbf841344b78b2e88700c8f44a25 : SUCCESS in 4m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/809cd333eacb4bbdaab4027d5c54acc1 : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/40ead8702a304dbc864665a1b39cfc6e : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c867ed1d313441b4a8c3b85a68d1f222 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2f4a031c63d44b5a3b911a9f01b88d5 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/145a36641ae24468987e92bfac4c8d14 : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a630db21d4174f7e849d9dc78c7dd5ba : SUCCESS in 6m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6446dc5849a646aa8e4f886f769ef50d : SUCCESS in 1h 59m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b7db271467648068db24a931dd0bc10 : SUCCESS in 1h 31m 03s","accounts_in_message":[],"_revision_number":4},{"id":"def072d68d368542d24fbdea44abc8f3f960ae0f","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-10-26 00:19:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4d3b4dbc61048828ba11d382ad8bf08e41c97ad9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-27 10:15:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9209bf3e8b908f8bcb54e4b3fb5e188bc3712c0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 10:15:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e2d9498b1519921d6c0cc14c563735c47880a5d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 11:49:56.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e11c58161c914f3dbb3c1d6a03bc3312 : SUCCESS in 5m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ee2defcb57e454688f81d595cea8a21 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6d5f00fd344b8f900ea573bd33513b : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45f4e2449bff435eb6cc6c8071ddc863 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e61c77785b524d08b6f8d19a714592fc : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9b74b5207d94c8384eafd405e3ccd52 : SUCCESS in 4m 35s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b47398639bdb4dc899b41902f44cad9a : SUCCESS in 1h 33m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f4603cec71c42fb96260ef3edcecda0 : SUCCESS in 1h 15m 52s","accounts_in_message":[],"_revision_number":4},{"id":"8b0af9bd21dca513a2f28333d4875c73f38c52df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 11:49:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"cc7c725b120c17feb457ef3da085077a36e91edb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 11:51:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ff50aa1a9854c0bb74f7be7c0304389 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b27c012972d8b4a0282e55573cc3aa6b8efd71eb","revisions":{"7b454aee48849b19f40dfd9e35dfaf844865f6d3":{"kind":"REWORK","_number":1,"created":"2019-10-02 07:50:52.000000000","uploader":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"ref":"refs/changes/97/686097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/97/686097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/1"}}},"commit":{"parents":[{"commit":"53f184739067739a0bb5a9dcbe2a69575fa380fd","subject":"Add missing DHCP port numbers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/53f184739067739a0bb5a9dcbe2a69575fa380fd"}]}],"author":{"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com","date":"2019-10-02 07:48:42.000000000","tz":120},"committer":{"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com","date":"2019-10-02 07:50:40.000000000","tz":120},"subject":"allow dns label to be equal to \u003ctenantid\u003e","message":"allow dns label to be equal to \u003ctenantid\u003e\n\nChange related : https://bugs.launchpad.net/neutron/+bug/1843218\n\nChange-Id: I6a1b983cf1bb0d6d04ec75b4725f39f29dc4c2ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b454aee48849b19f40dfd9e35dfaf844865f6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7b454aee48849b19f40dfd9e35dfaf844865f6d3"}]},"branch":"refs/heads/master"},"6414e0c2c370ce95e07fc05dd48cdb9bf155a859":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-03 07:45:15.000000000","uploader":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"ref":"refs/changes/97/686097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/97/686097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/2"}}},"commit":{"parents":[{"commit":"53f184739067739a0bb5a9dcbe2a69575fa380fd","subject":"Add missing DHCP port numbers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/53f184739067739a0bb5a9dcbe2a69575fa380fd"}]}],"author":{"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com","date":"2019-10-02 07:48:42.000000000","tz":120},"committer":{"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com","date":"2019-10-03 07:43:52.000000000","tz":120},"subject":"allow dns label to be equal to \u003ctenantid\u003e","message":"allow dns label to be equal to \u003ctenantid\u003e\n\nRelated-Bug: #1843218\nNeeded-By: https://review.opendev.org/#/c/686343/\n\nChange-Id: I6a1b983cf1bb0d6d04ec75b4725f39f29dc4c2ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6414e0c2c370ce95e07fc05dd48cdb9bf155a859"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/6414e0c2c370ce95e07fc05dd48cdb9bf155a859"}]},"branch":"refs/heads/master"},"153e59832f88282e27fc10aef82296e251d007b1":{"kind":"REWORK","_number":3,"created":"2019-10-21 06:43:51.000000000","uploader":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"ref":"refs/changes/97/686097/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/97/686097/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/3"}}},"commit":{"parents":[{"commit":"53f184739067739a0bb5a9dcbe2a69575fa380fd","subject":"Add missing DHCP port numbers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/53f184739067739a0bb5a9dcbe2a69575fa380fd"}]}],"author":{"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com","date":"2019-10-02 07:48:42.000000000","tz":120},"committer":{"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com","date":"2019-10-21 06:43:32.000000000","tz":120},"subject":"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label","message":"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label\n\nA DNS label is a alphanumerical text seperated by period(s).\n(i.e. label1.label2.label2.com.)\nHere we allow a label to be equal keywords defined which will be eval later\n\nRelated-Bug: #1843218\nNeeded-By: https://review.opendev.org/#/c/686343/\nChange-Id: I6a1b983cf1bb0d6d04ec75b4725f39f29dc4c2ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/153e59832f88282e27fc10aef82296e251d007b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/153e59832f88282e27fc10aef82296e251d007b1"}]},"branch":"refs/heads/master"},"b27c012972d8b4a0282e55573cc3aa6b8efd71eb":{"kind":"REWORK","_number":4,"created":"2019-10-23 06:00:00.000000000","uploader":{"_account_id":30900,"name":"Gregoire Mahe","email":"gregoire@mahe.lt","username":"gregoiremahe"},"ref":"refs/changes/97/686097/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/97/686097/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/97/686097/4"}}},"commit":{"parents":[{"commit":"b72d2d1b535058645287bd5e271a6d9f8325d285","subject":"Merge \"install neutron_lib international messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b72d2d1b535058645287bd5e271a6d9f8325d285"}]}],"author":{"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com","date":"2019-10-02 07:48:42.000000000","tz":120},"committer":{"name":"Gregoire Mahe","email":"gregoire.mahe@corp.ovh.com","date":"2019-10-23 05:59:44.000000000","tz":120},"subject":"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label","message":"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label\n\nA DNS label is a alphanumerical text seperated by period(s).\n(i.e. label1.label2.label2.com.)\nHere we allow a label to be equal keywords defined which will be eval later\n\nRelated-Bug: #1843218\nNeeded-By: https://review.opendev.org/#/c/686343/\nChange-Id: I6a1b983cf1bb0d6d04ec75b4725f39f29dc4c2ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b27c012972d8b4a0282e55573cc3aa6b8efd71eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b27c012972d8b4a0282e55573cc3aa6b8efd71eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
