)]}'
{"id":"openstack%2Fneutron-lib~689620","triplet_id":"openstack%2Fneutron-lib~stable%2Fstein~I76ab222da64589528defe434589b92a033ef8cdf","project":"openstack/neutron-lib","branch":"stable/stein","topic":"bug/1847586-stable/stein","hashtags":[],"change_id":"I76ab222da64589528defe434589b92a033ef8cdf","subject":"install neutron_lib international messages","status":"MERGED","created":"2019-10-19 11:27:23.000000000","updated":"2019-11-19 03:50:51.000000000","submitted":"2019-11-19 03:49:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689620-1574135381921-ca8dfbed","meta_rev_id":"ce00fe5590fa7e0c404d34f36de72a21d9520c27","_number":689620,"virtual_id_number":689620,"owner":{"_account_id":27079,"name":"Frank Wang","email":"wangpeihuixyz@126.com","username":"FrankWang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-19 03:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2019-11-18 10:20:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":2,"date":"2019-11-19 02:19:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"date":"2019-10-25 13:54:46.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2019-10-25 13:54:49.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"date":"2019-10-25 08:15:48.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":1,"date":"2019-11-19 02:19:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 08:15:48.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-10-25 13:54:46.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-10-25 13:54:49.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2019-11-18 10:20:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-11-19 02:19:07.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-11-19 03:49:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb042fec197127dc5a40d1799ebe4f9650afaa2f","author":{"_account_id":27079,"name":"Frank Wang","email":"wangpeihuixyz@126.com","username":"FrankWang"},"date":"2019-10-19 11:27:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0a41abdb1a43da79a60b60e9858a30288f167f8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-19 12:57:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47d35f3da4714e0b8a1de2eb91680366 : SUCCESS in 3m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e6f614a19ae4359b0ffa36380ab0dcb : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dad95f81f8494b86beb5789b36d46962 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f909881b9864d95a0987053e2657d21 : SUCCESS in 3m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d14449df1935442caf23a055aae46af4 : SUCCESS in 3m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c3e3d4b194a34be4a01cc673f9080a2e : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d16f22a41e60465a9633119512257669 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fabc1279c59d413f99bf1916f6dafab0 : SUCCESS in 3m 40s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cb6a4a31914c48ddad0ff870dc4da3df : SUCCESS in 1h 25m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c11da10bd5a644ac837e83bdda57ed6c : SUCCESS in 1h 12m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8a7cad744b8e79bc7619c4995e43dced601f79c3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-18 10:20:22.000000000","message":"Patch Set 1: Code-Review+2\n\nWe need to remember to release new release of neutron-lib for stable branch than. We didn\u0027t do this too often so far so lets be aware of this :)","accounts_in_message":[],"_revision_number":1},{"id":"232cc88c178bddd0df38c85293e4a3857f0b3e66","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-11-19 02:19:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd072a02a68d970942e6b292d35a9ee06291058d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 02:19:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b3ff055852d255b31a1ee194f4ba530ba306977","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 03:49:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1b1ec135315df2faf7ac752b1239e3f5f5e13652","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 03:49:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf48156252474f41977f727c57f47a63 : SUCCESS in 2m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4874f8b3f42f4a8d8766a8584218b458 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08e273a420ee48d982a4d3f03a24e9c8 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f06fafefee5e4644a51ef9724b01be99 : SUCCESS in 3m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/18f3a06071d541c8acff9c4ca391b86b : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0b82beb908e4a0992cae0dda3fc122e : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bdd69519ced444e9ccf25c42567b06d : SUCCESS in 3m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/51a4322b0bb444fdbd97293547411f92 : SUCCESS in 1h 26m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5eb202aab4b49cba0a515be4b8df50c : SUCCESS in 1h 12m 12s","accounts_in_message":[],"_revision_number":1},{"id":"3484ebaaef6d880c7e163587a0aff5a9f2babf2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 03:50:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1a2ae73b08c4f268a47b4e6c3cd0513 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3dcc921af53bc5012e3c89ff464084a8d0f3cf22","revisions":{"3dcc921af53bc5012e3c89ff464084a8d0f3cf22":{"kind":"REWORK","_number":1,"created":"2019-10-19 11:27:23.000000000","uploader":{"_account_id":27079,"name":"Frank Wang","email":"wangpeihuixyz@126.com","username":"FrankWang"},"ref":"refs/changes/20/689620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/20/689620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/689620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/689620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/20/689620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/20/689620/1"}}},"commit":{"parents":[{"commit":"bba52345ee219a85f41bc87ed7ff20f5de2d9b1c","subject":"Filter placement API endpoint by type too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/bba52345ee219a85f41bc87ed7ff20f5de2d9b1c"}]}],"author":{"name":"Frank Wang","email":"wangpeihuixyz@126.com","date":"2019-10-10 10:53:56.000000000","tz":480},"committer":{"name":"Frank Wang","email":"wangpeihuixyz@126.com","date":"2019-10-19 11:27:23.000000000","tz":0},"subject":"install neutron_lib international messages","message":"install neutron_lib international messages\n\nneutron-lib needs to load internationalized file at initialization,\notherwise, international logging message will not work. refer to [1]\nzh_CN neutron_lib.po has been completed.\n\n[1] https://review.opendev.org/#/c/536011/\n\nChange-Id: I76ab222da64589528defe434589b92a033ef8cdf\nCloses-Bug: #1847586\n(cherry picked from commit 66cbfc813ac5caa29b35b659e54935917471f00c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3dcc921af53bc5012e3c89ff464084a8d0f3cf22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/3dcc921af53bc5012e3c89ff464084a8d0f3cf22"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
