)]}'
{"id":"openstack%2Fneutron-lib~689945","triplet_id":"openstack%2Fneutron-lib~master~I9181332d560fdffa0e5398c14e41f77c1350c2e5","project":"openstack/neutron-lib","branch":"master","topic":"bulk_extend_resources","hashtags":[],"change_id":"I9181332d560fdffa0e5398c14e41f77c1350c2e5","subject":"Allow for both regular and bulk resource extenders","status":"ABANDONED","created":"2019-10-22 05:53:09.000000000","updated":"2021-05-21 13:23:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d61cc80da40e29bcce646b402534fb927503035e","_number":689945,"virtual_id_number":689945,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-22 07:45:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2021-05-21 13:23:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2019-10-23 07:56:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"date":"2019-10-23 07:56:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2019-10-23 14:39:24.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-22 05:55:27.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-10-22 07:45:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-23 07:56:31.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-10-23 07:56:36.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"1ba35f335ea9dd1c8c351d8b895e0ba8969ccd43","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-22 05:53:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d214afa2013902904b4556824385dc25e0a1e54e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2019-10-22 05:55:27.000000000","message":"Patch Set 1: Code-Review+1\n\nMaybe a release note for this?","accounts_in_message":[],"_revision_number":1},{"id":"41db3e414441866afdc3c5770fcfd6b2caea0229","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-22 07:32:40.000000000","message":"Patch Set 1:\n\n\u003e Maybe a release note for this?\n\nI thought about that, but this is still heavily POC and the details may change before the implementation is complete.  I would advocate in favor of the release note coming towards the end of the string of changes, since there will be a few more to implement this capability.","accounts_in_message":[],"_revision_number":1},{"id":"f979592b5a1ebdad86b4d4fadeb810c0ac509912","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-22 07:45:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f68f82180e144abb87742c7f1af672f4 : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d759ef501cb44b40b121e5ec7adf7551 : SUCCESS in 3m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f64cc42b12e4aeca2b31982c27c3dba : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63a536a9037941f68cacb680eaec2a05 : SUCCESS in 3m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/570d4008f22b4506b58dcdeaf73ad680 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/665d1c7ae6674f4880e87ea62068a821 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/752496339e364df397bbfef33195d776 : SUCCESS in 3m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dd922d35f23b416cabf725f821d0d661 : SUCCESS in 1h 50m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78e31ee4c0d04a74bad74116f7594fbb : SUCCESS in 1h 34m 02s","accounts_in_message":[],"_revision_number":1},{"id":"eacddac380e8e69fe2361675fb8f111fc021c3cd","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-23 14:39:24.000000000","message":"Patch Set 1: Workflow-1\n\nI need to figure out how to handle both a bulk and a singleton reservation here.  Let me take a hypothetical example - let\u0027s say a given resource extends port creation.  We might want both a singleton and a bulk version of that resource extender, so bulk-friendly functions could use the bulk-capable bulk extenders and then run a for loop over the ones that are not bulk firendly.  \n\nBut that founders on the shoals of identity - how can we tell that \"bulk extender function X\" is extending the same thing as \"singleton extender function Y\" when all that is saved is a weakref?","accounts_in_message":[],"_revision_number":1},{"id":"312271c5f6c1842283b59722b311f4bf1186960f","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-21 13:23:11.000000000","message":"Removed Code-Review+1 by Rodolfo Alonso \u003cralonsoh@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"d61cc80da40e29bcce646b402534fb927503035e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-05-21 13:23:46.000000000","message":"Abandoned\n\nIf needed, we can restore this patch and continue working on this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac7c39ae8e8dc4224d499bf075337ea4fd5d9744","revisions":{"ac7c39ae8e8dc4224d499bf075337ea4fd5d9744":{"kind":"REWORK","_number":1,"created":"2019-10-22 05:53:09.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/45/689945/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/45/689945/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/689945/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/689945/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/45/689945/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/45/689945/1"}}},"commit":{"parents":[{"commit":"b72d2d1b535058645287bd5e271a6d9f8325d285","subject":"Merge \"install neutron_lib international messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/b72d2d1b535058645287bd5e271a6d9f8325d285"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-10-22 05:50:21.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-10-22 05:53:07.000000000","tz":-240},"subject":"Allow for both regular and bulk resource extenders","message":"Allow for both regular and bulk resource extenders\n\nIn order to allow for the concept of a bulk resource_extend action, the\n_resource_extend_functions datastructure needs to accomodate the concept\nof a bulk resource.  This is accomplished by using the regular resource\nname with \"_bulk\" appended.\n\nThis change alters get_funcs to return registered resources that match\nthat pattern.\n\nChange-Id: I9181332d560fdffa0e5398c14e41f77c1350c2e5\nRelated-bug: #1843924\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ac7c39ae8e8dc4224d499bf075337ea4fd5d9744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/ac7c39ae8e8dc4224d499bf075337ea4fd5d9744"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
