)]}'
{"id":"openstack%2Fneutron-lib~690909","triplet_id":"openstack%2Fneutron-lib~master~I3ba42cf3ed54e49e8bc6a6deede398643dd1c78c","project":"openstack/neutron-lib","branch":"master","topic":"remove-external-net-limitations-leftovers","hashtags":[],"change_id":"I3ba42cf3ed54e49e8bc6a6deede398643dd1c78c","subject":"Remove TooManyExternalNetworks exception class","status":"MERGED","created":"2019-10-24 10:14:02.000000000","updated":"2020-03-13 20:49:29.000000000","submitted":"2020-03-13 20:48:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690909-1584132483982-cb471cbf","meta_rev_id":"863c968c789991fa1abc915ac4f26206e4f49226","_number":690909,"virtual_id_number":690909,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 20:48:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2020-02-28 10:59:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2020-03-10 22:49:29.000000000","_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2020-03-13 19:24:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-03-13 19:24:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-28 10:59:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2020-03-10 22:49:29.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2020-03-13 19:24:17.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-03-13 20:48:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"132eb5a7d78515a128bff19918945ccdb0411af6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-10-24 10:14:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"956ccf50ae84f47de1eff3c74574666b264b0a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 11:58:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863e58b91a9b447dbdb8c77772ae2b78 : SUCCESS in 7m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28a1b89d119f4231814979db537c520d : SUCCESS in 4m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c47f8299177488ea6e4ba1af92fe5eb : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00ffaad622aa438f9d1af08ffe68a0d8 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb5cc44001974a3d9c339d9e5b2e0667 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70d08528803d44e38fd18d046f4558f4 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd39b480cb01400284e7bf5a455d7a1f : SUCCESS in 5m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1e763284e28f47d7a51949506a53e6db : SUCCESS in 1h 41m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a949da589ae424f8a575d68aea36b2c : SUCCESS in 1h 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4a46f1861a000c67351dbf26f50693e413e4a4eb","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2019-10-24 19:29:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bf7d20236cd1b3a1b8fd20c4473639a7e46ce52","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-28 10:52:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"74a5e24981c8af1740e1a7f0232e77cddbf908be","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-28 10:53:16.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"94f055303c7e5383d55f2e58ce7b4ab4ed49d5fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 10:53:44.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"a06da863984e53588144704ed1c4935e9303f0bf","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-02-28 10:54:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"218ff51aa59c35bd72cad441eab6e0150f716a2f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-02-28 10:59:03.000000000","message":"Patch Set 3: Code-Review+1\n\nWaiting to https://review.opendev.org/#/c/710429 to merge","accounts_in_message":[],"_revision_number":3},{"id":"06babbff2caa06ddf159406bc6f990bb57bea1d0","author":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"date":"2020-03-10 22:49:29.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c276e98e65a221b9ec10e7e5862c5e14eef25f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 00:13:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed64caf6e3b4593ab4ce46982434c04 : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eced9155969241d3b3672838b5a0956b : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af22303ca8ea410497304bd18c970f6f : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf2948b8e9f14060afbb385f173c007c : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75e3e539d93b44c0bc3d15b38c7158a0 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2fc051a383a4c6792fd0c2a99af10e1 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0770c21954b243c880530acaea981edd : SUCCESS in 4m 13s (non-voting)\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/233d03a9bb3c4ce29c77da83edf66886 : SUCCESS in 1h 18m 28s","accounts_in_message":[],"_revision_number":3},{"id":"cbce9bf7cff23538be35460ae44ff3dc903f0f42","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2020-03-13 19:24:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3c5de5534c4d53605c36bc87536d6c8aeec6de98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 19:24:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6362900201eb2c82214adaf532472547838fbea0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 20:48:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a6861b9a3fe93fa58bcbea49f1411da77efc550b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 20:48:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f8c2f44fe14120a40faa61524815e3 : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8471a4867be04284bdfa60dd80ec75a4 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c270904cc84bbd9ed840fe49754d3a : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18da9a8ca0074e8d809a9f22d1bbf51e : SUCCESS in 3m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34c52df0572d46288045b7749eb01599 : SUCCESS in 5m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23ffb827f2414377b87643e6f34f4108 : SUCCESS in 1h 18m 32s","accounts_in_message":[],"_revision_number":3},{"id":"187dbc6c93099ee91f5053436d351c4ebc890450","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 20:49:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df80e4257be14185be5ffa8ff8d400b6 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"95bcd28daed4eaff999c08cf8e3d1e79ab851839","revisions":{"c3665125758dca151e8eee005de6e4b7f84054b7":{"kind":"REWORK","_number":1,"created":"2019-10-24 10:14:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/09/690909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/09/690909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/1"}}},"commit":{"parents":[{"commit":"81621f11b2a1e9800d3b0543b0e95dcb7a6aab3a","subject":"Merge \"Add retry and debug logs to NoAuthClient of placement client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/81621f11b2a1e9800d3b0543b0e95dcb7a6aab3a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-24 10:12:20.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-24 10:12:20.000000000","tz":120},"subject":"Remove TooManyExternalNetworks exception class","message":"Remove TooManyExternalNetworks exception class\n\nThis exception type is not used anymore as Neutron now don\u0027t have\n\u0027gateway_external_network_id\u0027 config option which was removed in\n[1].\n\n[1] https://review.opendev.org/#/c/666409/\n\nDepends-On: https://review.opendev.org/690908\nChange-Id: I3ba42cf3ed54e49e8bc6a6deede398643dd1c78c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c3665125758dca151e8eee005de6e4b7f84054b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/c3665125758dca151e8eee005de6e4b7f84054b7"}]},"branch":"refs/heads/master"},"e220603631e1e9755f57458178068f273bd0903d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-28 10:52:56.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/09/690909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/09/690909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/2"}}},"commit":{"parents":[{"commit":"33b41fb595e20fea3eb589307b81a116d27c067b","subject":"Merge \"Introduce rbac-address-scope api extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/33b41fb595e20fea3eb589307b81a116d27c067b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-24 10:12:20.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-28 10:52:56.000000000","tz":0},"subject":"Remove TooManyExternalNetworks exception class","message":"Remove TooManyExternalNetworks exception class\n\nThis exception type is not used anymore as Neutron now don\u0027t have\n\u0027gateway_external_network_id\u0027 config option which was removed in\n[1].\n\n[1] https://review.opendev.org/#/c/666409/\n\nDepends-On: https://review.opendev.org/690908\nChange-Id: I3ba42cf3ed54e49e8bc6a6deede398643dd1c78c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e220603631e1e9755f57458178068f273bd0903d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/e220603631e1e9755f57458178068f273bd0903d"}]},"branch":"refs/heads/master"},"95bcd28daed4eaff999c08cf8e3d1e79ab851839":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-28 10:53:16.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/09/690909/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/09/690909/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/09/690909/3"}}},"commit":{"parents":[{"commit":"33b41fb595e20fea3eb589307b81a116d27c067b","subject":"Merge \"Introduce rbac-address-scope api extension\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/33b41fb595e20fea3eb589307b81a116d27c067b"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2019-10-24 10:12:20.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-02-28 10:53:13.000000000","tz":0},"subject":"Remove TooManyExternalNetworks exception class","message":"Remove TooManyExternalNetworks exception class\n\nThis exception type is not used anymore as Neutron now don\u0027t have\n\u0027gateway_external_network_id\u0027 config option which was removed in\n[1].\n\n[1] https://review.opendev.org/#/c/666409/\n\nDepends-On: https://review.opendev.org/690908\nDepends-On: https://review.opendev.org/710429\nChange-Id: I3ba42cf3ed54e49e8bc6a6deede398643dd1c78c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/95bcd28daed4eaff999c08cf8e3d1e79ab851839"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/95bcd28daed4eaff999c08cf8e3d1e79ab851839"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
