)]}'
{"id":"openstack%2Fneutron-lib~692068","triplet_id":"openstack%2Fneutron-lib~master~Ibfc6dab327c9df59222cda3e6a1f5dc566aa1bde","project":"openstack/neutron-lib","branch":"master","topic":"bug/1850602","hashtags":[],"change_id":"Ibfc6dab327c9df59222cda3e6a1f5dc566aa1bde","subject":"Remove fwaas_v1 exceptions and definitions","status":"MERGED","created":"2019-10-30 08:01:38.000000000","updated":"2019-11-20 03:29:35.000000000","submitted":"2019-11-20 03:28:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692068-1574220502685-c4d0231e","meta_rev_id":"64b19eff15b32f9c3f4c9f460bf153a08c0a29bc","_number":692068,"virtual_id_number":692068,"owner":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2019-11-20 00:08:04.000000000","_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-20 03:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":2,"date":"2019-11-18 22:46:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2019-11-18 10:17:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2019-11-15 13:16:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":1,"date":"2019-11-18 22:46:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},{"value":0,"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 08:02:44.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2019-10-30 08:03:09.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2019-10-30 08:03:56.000000000","updated_by":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"reviewer":{"_account_id":27654,"name":"Hongbin Lu","email":"kira034@163.com","username":"hongbin.lu"},"state":"REVIEWER"},{"updated":"2019-10-30 08:04:31.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2019-10-30 08:59:47.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2019-10-30 14:12:31.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2019-11-15 13:16:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2019-11-18 10:17:22.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-11-18 22:46:05.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2019-11-20 03:28:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8bc8619754c01cfe235c37882059b1c35274b06","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-10-30 08:01:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1fbc7f20a412cebb131af7daa8077123784f4f5","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-10-30 08:52:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\njust one nit","accounts_in_message":[],"_revision_number":1},{"id":"333cdbf55ef664e59a8e919da297dc22d2d76a41","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-10-30 08:59:47.000000000","message":"Patch Set 1: Code-Review-1\n\nAs you may notice, this module is referred by neutron-fwaas code [1]\nI would suggest a final clean up of fwaas v1 code first.\n\n[1] https://opendev.org/openstack/neutron-fwaas/src/branch/master/neutron_fwaas/extensions/firewall.py#L23","accounts_in_message":[],"_revision_number":1},{"id":"d464b7a0f7793b2091ea65e8d30737ff377c5936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 09:37:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4220894491584dd0821dbc7f2c988e00 : SUCCESS in 6m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d00ae117a2f498c8361ca98ec04e64d : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87eb49094c7644edbfdebf0313ea83c4 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67d6d709a1004973b135f0cb6a0ade37 : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f54339e688a476c80fde942d251c159 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5bece9f3205645209c80fc1091e21261 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b536ca5c6ec74016bac346574d473efc : SUCCESS in 4m 02s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c7f8720c2bea47a3b954689f33e71386 : SUCCESS in 1h 34m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a0f2f24d5184144842ebdc0b3f6c204 : SUCCESS in 1h 15m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ee1cee3e0a3a0ad1448bfcd1a7bf80a2aad25fc1","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-10-30 13:28:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"85c982dc869d6d82fb725db547979990b2912643","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-10-30 13:29:06.000000000","message":"Patch Set 2:\n\nDone","accounts_in_message":[],"_revision_number":2},{"id":"f7ed4917202ca9783d4781c3164d167f260e2440","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-10-30 13:48:00.000000000","message":"Patch Set 2: Code-Review-1\n\nHi, Akihiro is right, neutron-fwaas depends on neutron-lib firewall_v1 definitions, as example see: https://opendev.org/openstack/neutron-fwaas/src/branch/master/neutron_fwaas/extensions/firewall.py#L23\n\ncheck with the Hound: http://codesearch.openstack.org/?q\u003dFirewallNotFound\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nSo please upload a patch that removes those dependencies from neutron-fwaas, and make this dependent on (with Depends-On:....) that patch.","accounts_in_message":[],"_revision_number":2},{"id":"03a607c89adce347c077203615f38bcd6cec0b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 13:55:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a90dcec3fd1249fd9a347c4a73f83a91 : SUCCESS in 5m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e70c30a29103492799ee50d9ac7ecb53 : FAILURE in 4m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efebda4aa3ae4c6baace42461ae32be0 : FAILURE in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/166da43cefef4c7d865f54f7683aa69d : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea81090f9f1f4e7180f3468fe2091203 : FAILURE in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a74ba39679134447a9ab002cc650a644 : FAILURE in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9db180966b742aa962d325f2a26b21c : FAILURE in 3m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/dfd03d616a5c43a2a1cf4242fd3f4263 : FAILURE in 24m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0df2e279d46346fd9c22869b4e2e6e0d : FAILURE in 23m 00s","accounts_in_message":[],"_revision_number":2},{"id":"190b6a852d03cc9d59805255e2bffcae4d4c9afb","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-10-30 14:12:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"96c0f08e7e0bc7b874cf34ae74d622e5ae5dc1ac","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-10-30 14:37:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c504d145dc6605847d2e16e52c20bd87bfb7ef3f","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-11-04 13:42:12.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0941fc72ca1e47ba9b956ea076b7be0e75082ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 14:06:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dff15e3b0334ab3a49e1f070fa91c7a : SUCCESS in 5m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26f71525c8f34b95a29a1a85f490a7e4 : FAILURE in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8966aec2cfb459eb067b6b8036fc77b : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ea994fbcb354debb683466f51d72a2d : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/753023be7aae4240ad1b8f2485126855 : FAILURE in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/102266eb213041dab419a4fac38a242a : FAILURE in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23cd45d4f6ff4d0c9c6419a2520626a0 : FAILURE in 4m 10s\n- tempest-full https://zuul.opendev.org/t/openstack/build/3dea2a56461a4045aa0ef228008e0c0b : FAILURE in 22m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ede9a4db2d64b2f972fe33e71e656b9 : FAILURE in 21m 20s","accounts_in_message":[],"_revision_number":3},{"id":"30707a7a0399804f637380f65221d43c6a3e59ab","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-11-04 14:51:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"00e377d08454958909def3434495ce565e229598","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-11-04 14:52:24.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"41a4de2c6ac9ebed6d2aa6ecc1fe78c03be414e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 16:20:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1fdc402d5b745d7afbba1b86673fbff : SUCCESS in 6m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a0aa3f89fef403ab14f8a2ef2db8be4 : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48891466a0774030a169ff2c81916f22 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8540fce29ab64f82b061a3fb3a7ef101 : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9fd205f2920f4080b3c86aaf89aec415 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4de0b7f0f3ae43238d401cee6e691dbf : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31f3b636418a43d98ff998950d5146d3 : SUCCESS in 3m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/11f38a0ef838406b8dfbdcb0b91f9e1b : SUCCESS in 1h 28m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f57496d2acd541f9821e08781212bc9d : SUCCESS in 1h 25m 11s","accounts_in_message":[],"_revision_number":4},{"id":"178480380e8b994def407213528621fb94eaa512","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-14 09:59:27.000000000","message":"Patch Set 4: Code-Review-1\n\nShould we also remove or at least add some note in firewall v1 api-ref: https://docs.openstack.org/api-ref/network/v2/#fwaas-v1-0-deprecated-fw-firewalls-firewall-policies-firewall-rules","accounts_in_message":[],"_revision_number":4},{"id":"eb4dc24661918bd03cbc006d53716496fd9c5cc5","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-11-14 12:11:01.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"18994da90152555eb277330693268786d16f0a01","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-11-14 12:12:06.000000000","message":"Patch Set 5:\n\nDone,remove fwaas_v1 api ref.","accounts_in_message":[],"_revision_number":5},{"id":"67c3649af33a97e302f2a8ad389d2efecadb81cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 13:55:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c94277255cfb475080d21820f42e0fe5 : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335f0cdb605146f1b09350324ce03e2b : SUCCESS in 4m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5c1da4442944a95944e2416f21cbf81 : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c478b67803aa4170bd86d6ffb690801d : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c05d6c3de9ab4607974f61139f56aafb : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a559786644cc4e52ba272a5d204a84ec : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9b400feaced41d39999213f1e0667a9 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7ee430ba9f664d32a9cb2e2177341ece : SUCCESS in 4m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0a0755ff804a4926997e86932a78850f : SUCCESS in 1h 30m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfdb150345624e1da6cd2fa2f16cec16 : SUCCESS in 1h 27m 15s","accounts_in_message":[],"_revision_number":5},{"id":"e1d166a37523703a26caf3b77b86c8116887131d","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-11-14 14:30:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0d7f664dab0003552cfca8a73bdbfcc2e13d33c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:47:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b88dd82719eb4fde96590d0e3df4a9a0 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26d73dbd79a047c6a0ffdebd20dd6420 : SUCCESS in 4m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d7aae531ff44c1fbee251aed9f1f75c : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f054d54808934e24870be09711cf5a72 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0898452fa4cd4496b35e3379e8fe432c : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6829bb181fc4e52aaa977eaa514f2dd : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49c7628a225a4115801e1d35f1b039f8 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6770d172998a4bd48383aeeb85c8e5d9 : FAILURE in 3m 14s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5de1559b154b4a65aeb1dcdf37e504a6 : FAILURE in 13m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edc03ec5f70040cf891371e4c535a137 : FAILURE in 13m 11s","accounts_in_message":[],"_revision_number":6},{"id":"9209ceee6d52e1a02efc1ab5281d40e8e87032de","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-11-14 15:11:59.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"45d2399a17e2f02914af3b4ff84a341c7a57f134","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 17:42:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1d16a21472d048e5a086998fb5b3b0a5 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559afc38398243a88bc87f5b8e1bb1ba : SUCCESS in 7m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70f4199d4ef54fd3a971d4fdac8b462c : SUCCESS in 7m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd64cb66688c4bf982f5980c313d5ac6 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f1ee06ff33d4ff3b791d776963f282a : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/046eff7f41c643f69d1372d98e0a8a7a : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a8b7f57a30e45b4930f756d797e56da : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/374f5e359a7943fd8081f4719a304882 : SUCCESS in 6m 00s\n- tempest-full https://zuul.opendev.org/t/openstack/build/64ffd1b91d2d4d40adfe7e17c1b59b1d : SUCCESS in 2h 17m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40ae74a31b92450a8bedb930dfdc995c : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":6},{"id":"95f129c00047f9d601dfdeb010c48c1ca7f56498","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2019-11-15 13:16:33.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"64397ad72b21c218b4fd9877e916799abcf2ed80","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-11-18 10:17:22.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"8b97b5822cbe987d3688e7f8158e37ce845af8e7","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2019-11-18 22:46:05.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"80b8b41fcfd808b26e3f4c32de1731c6cf0a5790","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-11-18 23:53:50.000000000","message":"Patch Set 6:\n\nPlease merge this patch(https://review.opendev.org/#/c/692094/) first, thanks.:)","accounts_in_message":[],"_revision_number":6},{"id":"f8dfb4593316bcadc84fa0059ce89f1013b45428","author":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"date":"2019-11-20 00:08:04.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"70529dedead19ee5ea9fce7bd61b6d95b20d8ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 01:49:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7d91507dedb44b55ba23e0080f98ec01 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cbc8d72c38433c8675056471e5c7cf : SUCCESS in 4m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39b298616244405880dd648776b5ac0e : SUCCESS in 3m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2600ff049fa14569bcd0775bafb9101b : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/998b6a7707f14177a2e62823ad4fa3b4 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5de1dda3c8584d4eae4539264b94cfc7 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb7b72aa4d66461caa9f106bd91d4533 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8bb59cdf920451ea404892bded2f008 : SUCCESS in 4m 09s\n- tempest-full https://zuul.opendev.org/t/openstack/build/432830927acf4d86a07d27b6ad8f1be7 : SUCCESS in 1h 36m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a27dfd25ae424064a40629efb2d003bb : SUCCESS in 1h 25m 29s","accounts_in_message":[],"_revision_number":6},{"id":"c82eb19df9fffda956b990379ca3bd3eb0368c86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 01:49:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fe6bf6ceed5db95c2b889a9cd4c1df612a7952c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 03:28:22.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/70e2b2d4bb5a484a85f56edd55805eee : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e118f82778a4ad6b4d55b6aed80d9b7 : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ecbb9ad41aa47fd83955d5ecbb2c495 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/403dfd314d074821b38e7c281a19f7fc : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ddb2730d1794068b54123c58795fcc4 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2604d1f4422428ea96c3ec7bd11c2f4 : SUCCESS in 3m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17e77732e35d4676a6f3679f2a4bfe8b : SUCCESS in 4m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/93eb13acf10745a48b07785f5838a861 : SUCCESS in 1h 37m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8fd4c35c0224c45b817aa36d7d986d4 : SUCCESS in 1h 23m 21s","accounts_in_message":[],"_revision_number":6},{"id":"8a423ab19dc578cdebf5791aeb6c43d0ddb6cb06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 03:28:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"3799787132f3f5b4fdf594c58b1799fdcaadfff2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 03:29:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/8218967c79484985a0fdec373af8872a : SUCCESS in 53s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc9ccfec77604bb3982b7adf9e444005 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9f56e984c3a00612cecc993f656187b13d93e88a","revisions":{"73f19370627e399d2d7b6d1c0d6d1c20454d904c":{"kind":"REWORK","_number":1,"created":"2019-10-30 08:01:38.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/68/692068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/68/692068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/1"}}},"commit":{"parents":[{"commit":"7e95f32e34e713046fbe9cc0118a88f221f57117","subject":"Merge \"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e95f32e34e713046fbe9cc0118a88f221f57117"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-10-30 07:51:31.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-10-30 08:01:35.000000000","tz":-240},"subject":"Remove fwaas_v1 exceptions","message":"Remove fwaas_v1 exceptions\n\nThere are still fwaas_v1 exceptions in neutron-lib, it\nis necessary to remove them because the fwaas_v1 code\nhas been removed in the Stein cycle.\n\nChange-Id: Ibfc6dab327c9df59222cda3e6a1f5dc566aa1bde\nCloses-bug: #1850602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/73f19370627e399d2d7b6d1c0d6d1c20454d904c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/73f19370627e399d2d7b6d1c0d6d1c20454d904c"}]},"branch":"refs/heads/master"},"f89801e0505e21097b9bdd0bbd30d71c71d2c6cc":{"kind":"REWORK","_number":2,"created":"2019-10-30 13:28:16.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/68/692068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/68/692068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/2"}}},"commit":{"parents":[{"commit":"7e95f32e34e713046fbe9cc0118a88f221f57117","subject":"Merge \"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e95f32e34e713046fbe9cc0118a88f221f57117"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-10-30 07:51:31.000000000","tz":-240},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-10-30 13:24:43.000000000","tz":-240},"subject":"Remove fwaas_v1 exceptions","message":"Remove fwaas_v1 exceptions\n\nThere are still fwaas_v1 exceptions in neutron-lib, it\nis necessary to remove them because the fwaas_v1 code\nhas been removed in the Stein cycle[1] and related\nextensions will be removed in patch[2].\n\n[1] https://review.opendev.org/#/c/616410/\n[2] https://review.opendev.org/#/c/692094/\nChange-Id: Ibfc6dab327c9df59222cda3e6a1f5dc566aa1bde\nCloses-bug: #1850602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f89801e0505e21097b9bdd0bbd30d71c71d2c6cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/f89801e0505e21097b9bdd0bbd30d71c71d2c6cc"}]},"branch":"refs/heads/master"},"23b57f97ff8ea31648b18704668cc9b14e5ae97c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-11-04 13:42:12.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/68/692068/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/68/692068/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/3"}}},"commit":{"parents":[{"commit":"7e95f32e34e713046fbe9cc0118a88f221f57117","subject":"Merge \"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e95f32e34e713046fbe9cc0118a88f221f57117"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-11-04 13:08:51.000000000","tz":-300},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-11-04 13:16:29.000000000","tz":-300},"subject":"Remove fwaas_v1 exceptions and definitions","message":"Remove fwaas_v1 exceptions and definitions\n\nThere are still fwaas_v1 exceptions and definitions in\nneutron-lib, it is necessary to remove them because the\nfwaas_v1 code has been removed in the Stein cycle[1] and\nthe related extensions will be removed.\n\n[1] https://review.opendev.org/#/c/616410/\nDepends-On: https://review.opendev.org/#/c/692094/\nChange-Id: Ibfc6dab327c9df59222cda3e6a1f5dc566aa1bde\nCloses-bug: #1850602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23b57f97ff8ea31648b18704668cc9b14e5ae97c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/23b57f97ff8ea31648b18704668cc9b14e5ae97c"}]},"branch":"refs/heads/master"},"351b8a1b2c10ae08489cdea9aa4ab18d37975373":{"kind":"REWORK","_number":4,"created":"2019-11-04 14:51:27.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/68/692068/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/68/692068/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/4"}}},"commit":{"parents":[{"commit":"7e95f32e34e713046fbe9cc0118a88f221f57117","subject":"Merge \"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e95f32e34e713046fbe9cc0118a88f221f57117"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-11-04 13:08:51.000000000","tz":-300},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-11-04 13:31:49.000000000","tz":-300},"subject":"Remove fwaas_v1 exceptions and definitions","message":"Remove fwaas_v1 exceptions and definitions\n\nThere are still fwaas_v1 exceptions and definitions in\nneutron-lib, it is necessary to remove them because the\nfwaas_v1 code has been removed in the Stein cycle[1] and\nthe related extensions will be removed.\n\n[1] https://review.opendev.org/#/c/616410/\nDepends-On: https://review.opendev.org/#/c/692094/\nChange-Id: Ibfc6dab327c9df59222cda3e6a1f5dc566aa1bde\nCloses-bug: #1850602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/351b8a1b2c10ae08489cdea9aa4ab18d37975373"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/351b8a1b2c10ae08489cdea9aa4ab18d37975373"}]},"branch":"refs/heads/master"},"dade8a999a20a01ed336c651633daa4b58f262b8":{"kind":"REWORK","_number":5,"created":"2019-11-14 12:11:01.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/68/692068/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/68/692068/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/5"}}},"commit":{"parents":[{"commit":"7e95f32e34e713046fbe9cc0118a88f221f57117","subject":"Merge \"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e95f32e34e713046fbe9cc0118a88f221f57117"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-11-04 13:08:51.000000000","tz":-300},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-11-04 13:47:56.000000000","tz":-300},"subject":"Remove fwaas_v1 exceptions and definitions","message":"Remove fwaas_v1 exceptions and definitions\n\nThere are still fwaas_v1 exceptions and definitions in\nneutron-lib, it is necessary to remove them because the\nfwaas_v1 code has been removed in the Stein cycle[1] and\nthe related extensions will be removed.\n\n[1] https://review.opendev.org/#/c/616410/\nDepends-On: https://review.opendev.org/#/c/692094/\nChange-Id: Ibfc6dab327c9df59222cda3e6a1f5dc566aa1bde\nCloses-bug: #1850602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dade8a999a20a01ed336c651633daa4b58f262b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/dade8a999a20a01ed336c651633daa4b58f262b8"}]},"branch":"refs/heads/master"},"9f56e984c3a00612cecc993f656187b13d93e88a":{"kind":"REWORK","_number":6,"created":"2019-11-14 14:30:44.000000000","uploader":{"_account_id":29157,"name":"zhanghao","email":"hao.zhang.am.i@gmail.com","username":"zhanghao2"},"ref":"refs/changes/68/692068/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/68/692068/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/68/692068/6"}}},"commit":{"parents":[{"commit":"7e95f32e34e713046fbe9cc0118a88f221f57117","subject":"Merge \"Allow \u003cproject_id\u003e, \u003cproject_name\u003e, \u003cuser_id\u003e and \u003cuser_name\u003e as DNS label\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/7e95f32e34e713046fbe9cc0118a88f221f57117"}]}],"author":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-11-04 13:08:51.000000000","tz":-300},"committer":{"name":"zhanghao","email":"zhanghao2@cmss.chinamobile.com","date":"2019-11-04 13:57:53.000000000","tz":-300},"subject":"Remove fwaas_v1 exceptions and definitions","message":"Remove fwaas_v1 exceptions and definitions\n\nThere are still fwaas_v1 exceptions and definitions in\nneutron-lib, it is necessary to remove them because the\nfwaas_v1 code has been removed in the Stein cycle[1] and\nthe related extensions will be removed.\n\n[1] https://review.opendev.org/#/c/616410/\nDepends-On: https://review.opendev.org/#/c/692094/\nChange-Id: Ibfc6dab327c9df59222cda3e6a1f5dc566aa1bde\nCloses-bug: #1850602\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9f56e984c3a00612cecc993f656187b13d93e88a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/9f56e984c3a00612cecc993f656187b13d93e88a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
