)]}'
{"id":"openstack%2Fneutron-lib~693396","triplet_id":"openstack%2Fneutron-lib~stable%2Frocky~Ic6eca08f2ccacb3f8bf741c47a45e88cd3877b29","project":"openstack/neutron-lib","branch":"stable/rocky","topic":"bug/1816485-stable/rocky","hashtags":[],"change_id":"Ic6eca08f2ccacb3f8bf741c47a45e88cd3877b29","subject":"Add setproctitle support to the workers module","status":"ABANDONED","created":"2019-11-07 16:59:51.000000000","updated":"2019-11-13 16:25:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4e8e580918aabd37305bb01744532e142b2158e1","_number":693396,"virtual_id_number":693396,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-07 21:41:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},{"value":-1,"date":"2019-11-13 13:43:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-07 16:59:51.000000000","updated_by":{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},"reviewer":{"_account_id":10980,"name":"Doug Wiegley","email":"dougwig@parkside.io","username":"dougw"},"state":"REVIEWER"},{"updated":"2019-11-07 21:41:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-13 13:43:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"cab1770c76c8fca43334dcb87baaa71893ea6860","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-07 16:59:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"62fb75ca6e9f48e4c07e307bb9ba4dac20c589dd","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-07 17:40:11.000000000","message":"Abandoned\n\nSorry, I thought this was a neutron change not a neutron-lib change.","accounts_in_message":[],"_revision_number":1},{"id":"b012afeb44c0fb3d122df827b754012d2a730a29","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-07 17:47:37.000000000","message":"Restored\n\nActually I really do want this.","accounts_in_message":[],"_revision_number":1},{"id":"7a936fef6935fa564c9216a7e4343ab45f0b7da0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 19:46:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a62e47e1a4004c409d1282709a267bb8 : FAILURE in 3m 16s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6d1b64a4c8f04dd4a34b9c318b70b98f : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec79addb79140939e6798955ce6a7b4 : FAILURE in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/580c64f594824fbeadf6505dae61cb0a : SUCCESS in 5m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/db7a1a110d904eee9b21e655266c71f5 : SUCCESS in 4m 57s\n- tempest-full https://zuul.opendev.org/t/openstack/build/9fe283e9d3d34a3983f723f8378ef03b : SUCCESS in 1h 52m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc8097b405b946e0ad43b8fec71d516b : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb46f97642a9481aa1b9e50063e3fbb4 : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f02c71ee5ac43428414c04af5c704c1 : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"608a4983775b88f1a9f34c6a98ec9e4a6e27be3d","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-07 20:17:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e25663b4c6e1cd49dc0731a3f3656a8123aec382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 21:41:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cca17066c6584a52bcb16c6617483231 : FAILURE in 2m 20s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dc97a3514c6b436b9a2b91a93af53f71 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c868a49e26a24e0fba4c8ba346501f5a : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae6b6e8b8b6c493691a531cf0bff0ff2 : SUCCESS in 3m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0ed005bce5aa41bb9b3c6c191ebaf0a4 : SUCCESS in 3m 34s\n- tempest-full https://zuul.opendev.org/t/openstack/build/42e44721a340480c9c66bffdfd83384c : SUCCESS in 1h 22m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c03fd07155a74303b38df265fc3ea518 : SUCCESS in 3m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1f04c19c91c4dbfa7688899e3c6f547 : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54a136e477354870858f8b3373565469 : SUCCESS in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"853e53515627582a96cfd0179e8de3c3cdd22be2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-11-13 13:43:16.000000000","message":"Patch Set 2: Code-Review-1\n\nWhile I love the change itself, I don\u0027t think we can backport this for:\n* added requirement\n* updated neutron needs this new neutron-lib, GA one will not work\n* user-visible change, we may break monitoring scripts (process name changing on minor update)\n\n(just commenting here, but applies to the cherry-picks set)","accounts_in_message":[],"_revision_number":2},{"id":"295418cb48ba26abdcf1ce10cbe6bf813941dc01","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2019-11-13 16:25:43.000000000","message":"Abandoned\n\nimpossible due to backport requirements change","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"013cbf6560f3eb9a2e4ff1a75ec63122835a6a5c","revisions":{"eca3db0d8f8097fb2bbe8a7f82e3a6053cec4c98":{"kind":"REWORK","_number":1,"created":"2019-11-07 16:59:51.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/96/693396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/96/693396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/693396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/693396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/693396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/96/693396/1"}}},"commit":{"parents":[{"commit":"48839690e1a4e6afcc1498441f19463bb8aef29a","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48839690e1a4e6afcc1498441f19463bb8aef29a"}]}],"author":{"name":"Doug Wiegley","email":"doug@parkside.io","date":"2019-02-15 03:24:02.000000000","tz":-420},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-11-07 16:59:51.000000000","tz":0},"subject":"Add setproctitle support to the workers module","message":"Add setproctitle support to the workers module\n\n- Set the process name of child neutron-servers to be something more\n  readily identifiable than today.\n- Enable by default for all users of the workers module, neutron will\n  have a conf setting for its workers.\n\nMatching neutron change:\nhttps://review.openstack.org/637019\n\nPartial-Bug: #1816485\nDepends-On: https://review.openstack.org/637024\nChange-Id: Ic6eca08f2ccacb3f8bf741c47a45e88cd3877b29\n(cherry picked from commit 2c1e09bfd5a275c74213d200a7455bbf46893a1a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eca3db0d8f8097fb2bbe8a7f82e3a6053cec4c98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/eca3db0d8f8097fb2bbe8a7f82e3a6053cec4c98"}]},"branch":"refs/heads/stable/rocky"},"013cbf6560f3eb9a2e4ff1a75ec63122835a6a5c":{"kind":"REWORK","_number":2,"created":"2019-11-07 20:17:46.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/96/693396/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron-lib","ref":"refs/changes/96/693396/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/693396/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/693396/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron-lib refs/changes/96/693396/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron-lib refs/changes/96/693396/2"}}},"commit":{"parents":[{"commit":"48839690e1a4e6afcc1498441f19463bb8aef29a","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/48839690e1a4e6afcc1498441f19463bb8aef29a"}]}],"author":{"name":"Doug Wiegley","email":"doug@parkside.io","date":"2019-02-15 03:24:02.000000000","tz":-420},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2019-11-07 20:17:45.000000000","tz":-300},"subject":"Add setproctitle support to the workers module","message":"Add setproctitle support to the workers module\n\n- Set the process name of child neutron-servers to be something more\n  readily identifiable than today.\n- Enable by default for all users of the workers module, neutron will\n  have a conf setting for its workers.\n\nMatching neutron change:\nhttps://review.openstack.org/637019\n\nPartial-Bug: #1816485\nDepends-On: https://review.openstack.org/637024\nChange-Id: Ic6eca08f2ccacb3f8bf741c47a45e88cd3877b29\n(cherry picked from commit 2c1e09bfd5a275c74213d200a7455bbf46893a1a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/013cbf6560f3eb9a2e4ff1a75ec63122835a6a5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron-lib/commit/013cbf6560f3eb9a2e4ff1a75ec63122835a6a5c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
